)]}'
{"id":"openstack%2Fproject-config~657061","triplet_id":"openstack%2Fproject-config~master~I2538a0fe22862f11aae18520708dc20701f8751d","project":"openstack/project-config","branch":"master","topic":"retire-mors","hashtags":[],"change_id":"I2538a0fe22862f11aae18520708dc20701f8751d","subject":"Retire mors","status":"ABANDONED","created":"2019-05-03 21:36:10.000000000","updated":"2019-05-16 12:52:52.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"1c406565b98856f2bf7e9eb068a7349c4a072164","_number":657061,"virtual_id_number":657061,"owner":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":1,"date":"2019-05-03 21:48:05.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-05-03 21:48:05.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"d41c1d63fae0b2bb5c6bd55e73ca3001b0f6c883","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2019-05-03 21:36:10.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"d7dc0d014eb789c6e7b31e6028c273f6a6eb6b44","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-03 21:48:05.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-zuul-jobs-linters http://logs.openstack.org/61/657061/1/check/openstack-zuul-jobs-linters/e7a84d2/ : SUCCESS in 6m 13s\n- project-config-gerrit http://logs.openstack.org/61/657061/1/check/project-config-gerrit/d9b09a9/ : SUCCESS in 3m 14s","accounts_in_message":[],"_revision_number":1},{"id":"4e3c6a4c40222adb79571169963e17d211720916","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2019-05-16 12:52:52.000000000","message":"Abandoned\n\nLooks like there\u0027s still like in this project yet [1]\n\n[1] https://review.opendev.org/#/c/657060/","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"3f4b11b775325167f9b5d33052a04cea5b976556","revisions":{"3f4b11b775325167f9b5d33052a04cea5b976556":{"kind":"REWORK","_number":1,"created":"2019-05-03 21:36:10.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/61/657061/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/61/657061/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/61/657061/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/61/657061/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/61/657061/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/61/657061/1"}}},"commit":{"parents":[{"commit":"945a586f65b31935a0359f286df5a2c155369412","subject":"Check groups entry in gerrit/projects","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/945a586f65b31935a0359f286df5a2c155369412"}]}],"author":{"name":"Stephen Finucane","email":"sfinucan@redhat.com","date":"2019-05-03 21:35:33.000000000","tz":-360},"committer":{"name":"Stephen Finucane","email":"sfinucan@redhat.com","date":"2019-05-03 21:35:33.000000000","tz":-360},"subject":"Retire mors","message":"Retire mors\n\nChange-Id: I2538a0fe22862f11aae18520708dc20701f8751d\nSigned-off-by: Stephen Finucane \u003csfinucan@redhat.com\u003e\nDepends-On: I183dc04a8bca3a1b954de1fe436c15af308dd278\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/3f4b11b775325167f9b5d33052a04cea5b976556"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/3f4b11b775325167f9b5d33052a04cea5b976556"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
