)]}'
{"id":"openstack%2Fproject-config~657316","triplet_id":"openstack%2Fproject-config~master~I006500a2c98f512176fbac4ab426a8719243fc70","project":"openstack/project-config","branch":"master","topic":"pip-no-editable","hashtags":[],"change_id":"I006500a2c98f512176fbac4ab426a8719243fc70","subject":"wheel-build: Work around pip 19.0 errors","status":"MERGED","created":"2019-05-06 05:45:30.000000000","updated":"2019-05-06 06:36:46.000000000","submitted":"2019-05-06 06:36:46.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"657316-1557124606647-959ddcf6","meta_rev_id":"34b3b0236b04bddf6624aaf376edee18c53e29f3","_number":657316,"virtual_id_number":657316,"owner":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":2,"date":"2019-05-06 06:36:46.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"all":[{"value":2,"date":"2019-05-06 05:49:41.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"all":[{"value":0,"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2019-05-06 06:25:05.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-05-06 05:49:41.000000000","updated_by":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"reviewer":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"state":"REVIEWER"},{"updated":"2019-05-06 06:36:46.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"c73fca2f783cd9266de825f738e7da0dad02769c","author":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"date":"2019-05-06 05:45:30.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"f58d577c3090da6e6abfc691ebfe472fd8e4caf8","author":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"date":"2019-05-06 05:46:12.000000000","message":"Uploaded patch set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"c05429351c637b2958041dbdc764828b2241f808","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2019-05-06 05:49:41.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"e13dfbad3300ceffd869cbf1471013b892f61551","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-06 05:52:49.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-zuul-jobs-linters http://logs.openstack.org/16/657316/2/check/openstack-zuul-jobs-linters/50959eb/ : SUCCESS in 5m 34s","accounts_in_message":[],"_revision_number":2},{"id":"2498b77b8bb73681d7a717e3aa89fc9da8c102d5","author":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"date":"2019-05-06 06:00:21.000000000","message":"Patch Set 2:\n\nI do believe this works, I replicated it with \"-e\" and then it installed without\n\n http://paste.openstack.org/show/750704/","accounts_in_message":[],"_revision_number":2},{"id":"4956b1a285000787c7c4f77b78df5f1adee872f6","author":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"date":"2019-05-06 06:25:05.000000000","message":"Patch Set 2: Workflow+1\n\nafter discussion i\u0027ll monitor a fresh job with this","accounts_in_message":[],"_revision_number":2},{"id":"1030197598ebf68bd544031e3ea8bc1a762e6c58","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-06 06:25:18.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"5ec53dc174e9c2630d4a711a580fa2dd911b81a0","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-06 06:36:46.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"7fd19e1cc9e93b693adbc75bac372aaaceebc67f","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-06 06:36:46.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-zuul-jobs-linters http://logs.openstack.org/16/657316/2/gate/openstack-zuul-jobs-linters/52eab7a/ : SUCCESS in 6m 16s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"cd3894ea51be03f3b236aa8d2ea2f27b6ff5c57b","revisions":{"61366a7ecc65e5b11914e40c8bfa3814d8452864":{"kind":"REWORK","_number":1,"created":"2019-05-06 05:45:30.000000000","uploader":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"ref":"refs/changes/16/657316/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/16/657316/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/16/657316/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/16/657316/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/16/657316/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/16/657316/1"}}},"commit":{"parents":[{"commit":"945a586f65b31935a0359f286df5a2c155369412","subject":"Check groups entry in gerrit/projects","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/945a586f65b31935a0359f286df5a2c155369412"}]}],"author":{"name":"Ian Wienand","email":"iwienand@redhat.com","date":"2019-05-06 05:34:57.000000000","tz":600},"committer":{"name":"Ian Wienand","email":"iwienand@redhat.com","date":"2019-05-06 05:34:57.000000000","tz":600},"subject":"Work around pip errors","message":"Work around pip errors\n\nAs pointed out by ajaeger; the wheel build jobs are failing with\n\n Error installing \u0027pip from\n git+https://github.com/ianw/pip.git@path-and-hash#egg\u003dpip\u0027: editable\n mode is not supported for pyproject.toml-style projects.\n\nThis appears to be an issue with pip 19.0 which would be installed on\nthe build hosts, interacting with projects that have a pyproject.toml\nfile.\n\nThe problem only seems to be related to installing in editable mode,\nwhich we don\u0027t really need here on this ephemeral host.  So just\nswitch to a regular install, with upgrade/reinstall to be sure.\n\n[1] https://github.com/pypa/pip/pull/6449\n\nChange-Id: I006500a2c98f512176fbac4ab426a8719243fc70\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/61366a7ecc65e5b11914e40c8bfa3814d8452864"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/61366a7ecc65e5b11914e40c8bfa3814d8452864"}]},"branch":"refs/heads/master"},"cd3894ea51be03f3b236aa8d2ea2f27b6ff5c57b":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2019-05-06 05:46:12.000000000","uploader":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"ref":"refs/changes/16/657316/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/16/657316/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/16/657316/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/16/657316/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/16/657316/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/16/657316/2"}}},"commit":{"parents":[{"commit":"945a586f65b31935a0359f286df5a2c155369412","subject":"Check groups entry in gerrit/projects","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/945a586f65b31935a0359f286df5a2c155369412"}]}],"author":{"name":"Ian Wienand","email":"iwienand@redhat.com","date":"2019-05-06 05:34:57.000000000","tz":600},"committer":{"name":"Ian Wienand","email":"iwienand@redhat.com","date":"2019-05-06 05:45:49.000000000","tz":600},"subject":"wheel-build: Work around pip 19.0 errors","message":"wheel-build: Work around pip 19.0 errors\n\nAs pointed out by ajaeger; the wheel build jobs are failing with\n\n Error installing \u0027pip from\n git+https://github.com/ianw/pip.git@path-and-hash#egg\u003dpip\u0027: editable\n mode is not supported for pyproject.toml-style projects.\n\nThis appears to be an issue with pip 19.0 which would be installed on\nthe build hosts, interacting with projects that have a pyproject.toml\nfile.\n\nThe problem only seems to be related to installing in editable mode,\nwhich we don\u0027t really need here on this ephemeral host.  So just\nswitch to a regular install, with upgrade/reinstall to be sure.\n\n[1] https://github.com/pypa/pip/pull/6449\n\nChange-Id: I006500a2c98f512176fbac4ab426a8719243fc70\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/cd3894ea51be03f3b236aa8d2ea2f27b6ff5c57b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/cd3894ea51be03f3b236aa8d2ea2f27b6ff5c57b"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
