)]}'
{"id":"openstack%2Fproject-config~657657","triplet_id":"openstack%2Fproject-config~master~Ib75714649848538e9fed171abd0b11f6fbc55503","project":"openstack/project-config","branch":"master","hashtags":[],"change_id":"Ib75714649848538e9fed171abd0b11f6fbc55503","subject":"Create \u0027Backport-Candidate\u0027 for Octavia repos","status":"MERGED","created":"2019-05-07 17:57:05.000000000","updated":"2019-05-23 21:28:22.000000000","submitted":"2019-05-23 21:28:22.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":3,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"657657-1558646902827-5a0d4aca","meta_rev_id":"239972b1fc3a1a3279818776af9fa6760b5942ce","_number":657657,"virtual_id_number":657657,"owner":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"value":0,"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"value":0,"_account_id":8099,"name":"Graham Hayes","email":"gr@ham.ie","username":"graham"},{"value":0,"_account_id":1955,"name":"Alan Pevec","email":"alan.pevec@redhat.com","username":"apevec"},{"value":0,"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":0,"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":0,"_account_id":28502,"name":"Colin Gibbons","email":"cgibbons@blizzard.com","username":"colin"},{"value":0,"date":"2019-05-23 17:15:25.000000000","_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":2,"date":"2019-05-23 21:28:22.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":28543,"name":"shupeng","email":"15050873171@163.com","username":"Vieri"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"recommended":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":1,"date":"2019-05-07 18:35:11.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"value":2,"date":"2019-05-23 21:19:29.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"value":1,"date":"2019-05-08 12:07:21.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":8099,"name":"Graham Hayes","email":"gr@ham.ie","username":"graham"},{"value":0,"_account_id":1955,"name":"Alan Pevec","email":"alan.pevec@redhat.com","username":"apevec"},{"value":2,"date":"2019-05-14 04:56:38.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":1,"date":"2019-05-07 18:28:08.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":1,"date":"2019-05-07 18:37:41.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":28502,"name":"Colin Gibbons","email":"cgibbons@blizzard.com","username":"colin"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2019-05-14 08:01:47.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":28543,"name":"shupeng","email":"15050873171@163.com","username":"Vieri"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"value":1,"date":"2019-05-23 21:19:29.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"value":0,"_account_id":8099,"name":"Graham Hayes","email":"gr@ham.ie","username":"graham"},{"value":0,"date":"2019-05-08 07:57:10.000000000","_account_id":1955,"name":"Alan Pevec","email":"alan.pevec@redhat.com","username":"apevec"},{"value":0,"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":0,"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":0,"_account_id":28502,"name":"Colin Gibbons","email":"cgibbons@blizzard.com","username":"colin"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":28543,"name":"shupeng","email":"15050873171@163.com","username":"Vieri"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1955,"name":"Alan Pevec","email":"alan.pevec@redhat.com","username":"apevec"},{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"_account_id":8099,"name":"Graham Hayes","email":"gr@ham.ie","username":"graham"},{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28502,"name":"Colin Gibbons","email":"cgibbons@blizzard.com","username":"colin"},{"_account_id":28543,"name":"shupeng","email":"15050873171@163.com","username":"Vieri"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-05-07 18:02:39.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"},{"updated":"2019-05-07 18:28:08.000000000","updated_by":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"reviewer":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"state":"REVIEWER"},{"updated":"2019-05-07 18:35:11.000000000","updated_by":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"reviewer":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"state":"REVIEWER"},{"updated":"2019-05-07 18:37:41.000000000","updated_by":{"_account_id":28502,"name":"Colin Gibbons","email":"cgibbons@blizzard.com","username":"colin"},"reviewer":{"_account_id":28502,"name":"Colin Gibbons","email":"cgibbons@blizzard.com","username":"colin"},"state":"REVIEWER"},{"updated":"2019-05-08 07:57:10.000000000","updated_by":{"_account_id":1955,"name":"Alan Pevec","email":"alan.pevec@redhat.com","username":"apevec"},"reviewer":{"_account_id":1955,"name":"Alan Pevec","email":"alan.pevec@redhat.com","username":"apevec"},"state":"REVIEWER"},{"updated":"2019-05-08 12:07:21.000000000","updated_by":{"_account_id":8099,"name":"Graham Hayes","email":"gr@ham.ie","username":"graham"},"reviewer":{"_account_id":8099,"name":"Graham Hayes","email":"gr@ham.ie","username":"graham"},"state":"REVIEWER"},{"updated":"2019-05-14 04:56:38.000000000","updated_by":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"reviewer":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"state":"REVIEWER"},{"updated":"2019-05-14 08:01:47.000000000","updated_by":{"_account_id":28543,"name":"shupeng","email":"15050873171@163.com","username":"Vieri"},"reviewer":{"_account_id":28543,"name":"shupeng","email":"15050873171@163.com","username":"Vieri"},"state":"REVIEWER"},{"updated":"2019-05-23 21:19:29.000000000","updated_by":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"reviewer":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"state":"REVIEWER"},{"updated":"2019-05-23 21:28:22.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"ea3402c197ec018e64ea93a633470fa603053389","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2019-05-07 17:57:05.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"293adb84d2e6df21f406490850fe87ac18fae11b","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2019-05-07 17:58:07.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"a8af31f2cd7ac1ae075f31e09100af7ecfbb6a03","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2019-05-07 18:01:54.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"6e1b12746a496f6243bbcc88b5f1d40c99a6742a","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2019-05-07 18:02:39.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"c3aebb8561f3d6f63a477bd9c9eec2a6ec6033c4","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-07 18:10:35.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-zuul-jobs-linters http://logs.openstack.org/57/657657/1/check/openstack-zuul-jobs-linters/4972984/ : SUCCESS in 7m 08s\n- project-config-gerrit http://logs.openstack.org/57/657657/1/check/project-config-gerrit/66111a2/ : FAILURE in 3m 37s","accounts_in_message":[],"_revision_number":1},{"id":"1d43b0d3fb2be42f5d868a9e94581642498152d1","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2019-05-07 18:16:51.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"983707deca247759f95219c024eef5bc68837b46","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-07 18:26:50.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-zuul-jobs-linters http://logs.openstack.org/57/657657/2/check/openstack-zuul-jobs-linters/f6a61cc/ : SUCCESS in 7m 32s\n- project-config-gerrit http://logs.openstack.org/57/657657/2/check/project-config-gerrit/78958ce/ : SUCCESS in 3m 18s","accounts_in_message":[],"_revision_number":2},{"id":"236a756d96744fbed66116b06d0e1528d747ee71","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2019-05-07 18:28:08.000000000","message":"Patch Set 2: Code-Review+1\n\nLGTM. Thank you!","accounts_in_message":[],"_revision_number":2},{"id":"4c476b700a21a809ab5ee48701aef0f03595ef01","author":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"date":"2019-05-07 18:35:11.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"36dd936180aae4a69542f8ec654660ab5a43fe7f","author":{"_account_id":28502,"name":"Colin Gibbons","email":"cgibbons@blizzard.com","username":"colin"},"date":"2019-05-07 18:37:41.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"9e5c296b0e4e069582e14b2d318b81b6a3da50f0","author":{"_account_id":8099,"name":"Graham Hayes","email":"gr@ham.ie","username":"graham"},"date":"2019-05-08 12:07:21.000000000","message":"Patch Set 2: Code-Review+1\n\nLooks correct.\n\nShould this copy +/- 2 votes? - Right now votes will be lost on a non trivial update","accounts_in_message":[],"_revision_number":2},{"id":"bd789967d15053b924af3ad61b75a0da51c91c2b","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2019-05-08 13:44:40.000000000","message":"Patch Set 2:\n\nGraham, wouldn\u0027t that be \"copyAllScoresOnTrivialRebase \u003d true\"? I see that set in all four project config files.","accounts_in_message":[],"_revision_number":2},{"id":"e2d5976335d64fbd90593e665a19f77ad9868a08","author":{"_account_id":8099,"name":"Graham Hayes","email":"gr@ham.ie","username":"graham"},"date":"2019-05-08 13:53:46.000000000","message":"Patch Set 2:\n\nit would be \n\n    copyMaxScore \u003d true\n    copyMinScore \u003d true\n\nto copy +/- 2 in all cases (not just on a rebase or commit message update)","accounts_in_message":[],"_revision_number":2},{"id":"546384a68c5cbd6e8743c6673b498d76587903e2","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2019-05-08 15:15:35.000000000","message":"Patch Set 2:\n\nWe wanted it to reset to 0 if there is a code change to be re-reviewed for backport compatibility.\nI.e. we don\u0027t want someone sneaking a change in that does not follow stable policy.","accounts_in_message":[],"_revision_number":2},{"id":"2a75bc632e83a7f9e77c515945f4d931fa616925","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2019-05-14 04:56:38.000000000","message":"Patch Set 2: Code-Review+2\n\nUnusual - let\u0027s experiment ;)","accounts_in_message":[],"_revision_number":2},{"id":"7eee78f97582c595110545e523244d47d6343411","author":{"_account_id":28543,"name":"shupeng","email":"15050873171@163.com","username":"Vieri"},"date":"2019-05-14 08:01:47.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"e5751d1e0284fcb35f73a11f05d93b913ee92e03","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2019-05-23 17:15:25.000000000","message":"Patch Set 2:\n\nbumping this...","accounts_in_message":[],"_revision_number":2},{"id":"313b374ddcba15fb9ccd43a9ab7a4821caa07e74","author":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"date":"2019-05-23 21:19:29.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"dec487f16bb6e210407677703f06d63e95c75cc1","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-23 21:19:46.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"62b43d50ef64847f5ca10c0fd3144c9783fc6d3a","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-23 21:28:22.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"239972b1fc3a1a3279818776af9fa6760b5942ce","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-23 21:28:22.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-zuul-jobs-linters http://logs.openstack.org/57/657657/2/gate/openstack-zuul-jobs-linters/0b28cb2/ : SUCCESS in 6m 02s\n- project-config-gerrit http://logs.openstack.org/57/657657/2/gate/project-config-gerrit/8670a2b/ : SUCCESS in 3m 08s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"c49fb365ff65178e928cdd021521239235516044","revisions":{"f751c90b08bcda560fd8823992b1ff2d83791eb5":{"kind":"REWORK","_number":1,"created":"2019-05-07 17:57:05.000000000","uploader":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"ref":"refs/changes/57/657657/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/57/657657/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/57/657657/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/57/657657/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/57/657657/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/57/657657/1"}}},"commit":{"parents":[{"commit":"159f7cd985148344a7e54775cd9642e6f7311d63","subject":"openafs-client : update kdc servers","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/159f7cd985148344a7e54775cd9642e6f7311d63"}]}],"author":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2019-05-07 17:54:06.000000000","tz":-420},"committer":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2019-05-07 17:57:03.000000000","tz":-420},"subject":"Create \u0027Backport-Candidate\u0027 for Octavia repos","message":"Create \u0027Backport-Candidate\u0027 for Octavia repos\n\nThe Octavia team would like to enabled passive voting on patches\nfor backport candidates. This means that backport candidate votes\nwill not block a patch from merging, but will allow the team to\nbetter track patches that should be backported.\n\nChange-Id: Ib75714649848538e9fed171abd0b11f6fbc55503\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/f751c90b08bcda560fd8823992b1ff2d83791eb5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/f751c90b08bcda560fd8823992b1ff2d83791eb5"}]},"branch":"refs/heads/master"},"c49fb365ff65178e928cdd021521239235516044":{"kind":"REWORK","_number":2,"created":"2019-05-07 18:16:51.000000000","uploader":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"ref":"refs/changes/57/657657/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/57/657657/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/57/657657/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/57/657657/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/57/657657/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/57/657657/2"}}},"commit":{"parents":[{"commit":"159f7cd985148344a7e54775cd9642e6f7311d63","subject":"openafs-client : update kdc servers","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/159f7cd985148344a7e54775cd9642e6f7311d63"}]}],"author":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2019-05-07 17:54:06.000000000","tz":-420},"committer":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2019-05-07 18:16:21.000000000","tz":-420},"subject":"Create \u0027Backport-Candidate\u0027 for Octavia repos","message":"Create \u0027Backport-Candidate\u0027 for Octavia repos\n\nThe Octavia team would like to enabled passive voting on patches\nfor backport candidates. This means that backport candidate votes\nwill not block a patch from merging, but will allow the team to\nbetter track patches that should be backported.\n\nChange-Id: Ib75714649848538e9fed171abd0b11f6fbc55503\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/c49fb365ff65178e928cdd021521239235516044"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/c49fb365ff65178e928cdd021521239235516044"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
