)]}'
{"id":"openstack%2Fproject-config~657853","triplet_id":"openstack%2Fproject-config~master~If5ad906de60035a0561018b6627ee445fce9ba76","project":"openstack/project-config","branch":"master","hashtags":[],"change_id":"If5ad906de60035a0561018b6627ee445fce9ba76","subject":"Refactor dashboards for neutron-dynamic-routing","status":"ABANDONED","created":"2019-05-08 17:22:44.000000000","updated":"2019-05-30 21:36:25.000000000","total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"44c44b7b1b9059cc3037684a43de984b0f7810d3","_number":657853,"virtual_id_number":657853,"owner":{"_account_id":4187,"name":"Ryan Tidwell","email":"rtidwell@suse.com","username":"ryan-tidwell"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},{"_account_id":4187,"name":"Ryan Tidwell","email":"rtidwell@suse.com","username":"ryan-tidwell"},{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":-1,"date":"2019-05-08 21:50:21.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":4187,"name":"Ryan Tidwell","email":"rtidwell@suse.com","username":"ryan-tidwell"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"rejected":{"_account_id":4187,"name":"Ryan Tidwell","email":"rtidwell@suse.com","username":"ryan-tidwell"},"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},{"value":-1,"date":"2019-05-14 14:14:17.000000000","permitted_voting_range":{"min":-1,"max":0},"_account_id":4187,"name":"Ryan Tidwell","email":"rtidwell@suse.com","username":"ryan-tidwell"},{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"_account_id":4187,"name":"Ryan Tidwell","email":"rtidwell@suse.com","username":"ryan-tidwell"},{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-05-08 17:22:58.000000000","updated_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"reviewer":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"state":"REVIEWER"},{"updated":"2019-05-08 17:23:10.000000000","updated_by":{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},"reviewer":{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},"state":"REVIEWER"},{"updated":"2019-05-08 17:28:02.000000000","updated_by":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"reviewer":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"state":"REVIEWER"},{"updated":"2019-05-08 21:50:21.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"b2026c30364bd754cdf0ba06bcd0dc02297fd012","author":{"_account_id":4187,"name":"Ryan Tidwell","email":"rtidwell@suse.com","username":"ryan-tidwell"},"date":"2019-05-08 17:22:44.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"d5eb2ca9d46e045e73ac1d9b4649ccd3a03d14b5","author":{"_account_id":4187,"name":"Ryan Tidwell","email":"rtidwell@suse.com","username":"ryan-tidwell"},"date":"2019-05-08 17:25:23.000000000","message":"Patch Set 1: Workflow-1\n\nI blindly overwrote the existing config in an attempt to use neutron as an example. It wasn\u0027t obvious until looking at the diff. This needs to be re-worked.","accounts_in_message":[],"_revision_number":1},{"id":"c2ac55d2c7f1814e60acf11dfd7381eb52ff0c44","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2019-05-08 17:28:02.000000000","message":"Patch Set 1:\n\n(1 comment)\n\nI suggest you merge https://review.opendev.org/#/c/652503/ and remove the py35 jobs now instead of another change in a few days ;)","accounts_in_message":[],"_revision_number":1},{"id":"74a5446b9f6c73c7ed64e1a789e94222ac2d448c","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-08 17:30:14.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-zuul-jobs-linters http://logs.openstack.org/53/657853/1/check/openstack-zuul-jobs-linters/e855d3c/ : SUCCESS in 6m 11s\n- project-config-grafyaml http://logs.openstack.org/53/657853/1/check/project-config-grafyaml/84acc66/ : SUCCESS in 3m 01s","accounts_in_message":[],"_revision_number":1},{"id":"ea7750825f1330236335d421bd0bb204343f14f8","author":{"_account_id":4187,"name":"Ryan Tidwell","email":"rtidwell@suse.com","username":"ryan-tidwell"},"date":"2019-05-08 21:44:00.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"99607ec3e07ea38a9a6a587ff23ee47fd3e4de40","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-08 21:50:21.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-zuul-jobs-linters http://logs.openstack.org/53/657853/2/check/openstack-zuul-jobs-linters/3b86e0e/ : SUCCESS in 5m 23s\n- project-config-grafyaml http://logs.openstack.org/53/657853/2/check/project-config-grafyaml/26fecc2/ : FAILURE in 2m 24s","accounts_in_message":[],"_revision_number":2},{"id":"20faa522132b0e2883f5b6f7f4d1e2e0cf81148b","author":{"_account_id":4187,"name":"Ryan Tidwell","email":"rtidwell@suse.com","username":"ryan-tidwell"},"date":"2019-05-14 14:14:17.000000000","message":"Patch Set 2: Workflow-1","accounts_in_message":[],"_revision_number":2},{"id":"44c44b7b1b9059cc3037684a43de984b0f7810d3","author":{"_account_id":4187,"name":"Ryan Tidwell","email":"rtidwell@suse.com","username":"ryan-tidwell"},"date":"2019-05-30 21:36:25.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"a6689cbdcd6853c6f62a866980018f584070adbe","revisions":{"bd5f2d92f0c72febff417ada5ea9f4d08a271097":{"kind":"REWORK","_number":1,"created":"2019-05-08 17:22:44.000000000","uploader":{"_account_id":4187,"name":"Ryan Tidwell","email":"rtidwell@suse.com","username":"ryan-tidwell"},"ref":"refs/changes/53/657853/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/53/657853/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/53/657853/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/53/657853/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/53/657853/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/53/657853/1"}}},"commit":{"parents":[{"commit":"159f7cd985148344a7e54775cd9642e6f7311d63","subject":"openafs-client : update kdc servers","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/159f7cd985148344a7e54775cd9642e6f7311d63"}]}],"author":{"name":"Ryan Tidwell","email":"rtidwell@suse.com","date":"2019-05-08 17:21:34.000000000","tz":-300},"committer":{"name":"Ryan Tidwell","email":"rtidwell@suse.com","date":"2019-05-08 17:22:43.000000000","tz":-300},"subject":"Add dashboards for neutron-dynamic-routing","message":"Add dashboards for neutron-dynamic-routing\n\nThis changes adds grafana dashboards for tracking the status of\nneutron-dynamic-routing jobs.\n\nChange-Id: If5ad906de60035a0561018b6627ee445fce9ba76\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/bd5f2d92f0c72febff417ada5ea9f4d08a271097"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/bd5f2d92f0c72febff417ada5ea9f4d08a271097"}]},"branch":"refs/heads/master"},"a6689cbdcd6853c6f62a866980018f584070adbe":{"kind":"REWORK","_number":2,"created":"2019-05-08 21:44:00.000000000","uploader":{"_account_id":4187,"name":"Ryan Tidwell","email":"rtidwell@suse.com","username":"ryan-tidwell"},"ref":"refs/changes/53/657853/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/53/657853/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/53/657853/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/53/657853/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/53/657853/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/53/657853/2"}}},"commit":{"parents":[{"commit":"159f7cd985148344a7e54775cd9642e6f7311d63","subject":"openafs-client : update kdc servers","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/159f7cd985148344a7e54775cd9642e6f7311d63"}]}],"author":{"name":"Ryan Tidwell","email":"rtidwell@suse.com","date":"2019-05-08 17:21:34.000000000","tz":-300},"committer":{"name":"Ryan Tidwell","email":"rtidwell@suse.com","date":"2019-05-08 21:42:34.000000000","tz":-300},"subject":"Refactor dashboards for neutron-dynamic-routing","message":"Refactor dashboards for neutron-dynamic-routing\n\nThis changes adjusts grafana dashboards for tracking the status of\nneutron-dynamic-routing jobs, providing improved organization and\ndropping tracking of python 3.5 jobs.\n\nChange-Id: If5ad906de60035a0561018b6627ee445fce9ba76\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/a6689cbdcd6853c6f62a866980018f584070adbe"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/a6689cbdcd6853c6f62a866980018f584070adbe"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
