)]}'
{"id":"openstack%2Fproject-config~657862","triplet_id":"openstack%2Fproject-config~master~I1e67a913ac61ef78b9a57d564f3e50bbe2fe4a45","project":"openstack/project-config","branch":"master","hashtags":[],"change_id":"I1e67a913ac61ef78b9a57d564f3e50bbe2fe4a45","subject":"Add spectroscope repository","status":"MERGED","created":"2019-05-08 18:29:46.000000000","updated":"2019-05-17 18:48:48.000000000","submitted":"2019-05-17 18:48:48.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":3,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"657862-1558118928784-d1a6a7d1","meta_rev_id":"be0c1eaabf79b4a423e7404c7204ac8dafbe38d5","_number":657862,"virtual_id_number":657862,"owner":{"_account_id":2903,"name":"Morgan Fainberg","email":"morgan.fainberg@gmail.com","username":"mdrnstm"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":16465,"name":"Kristi Nikolla","email":"knikolla@bu.edu","username":"knikolla"},{"value":0,"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"value":0,"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":0,"date":"2019-05-11 18:13:22.000000000","_account_id":2903,"name":"Morgan Fainberg","email":"morgan.fainberg@gmail.com","username":"mdrnstm"},{"value":0,"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},{"value":2,"date":"2019-05-17 18:48:48.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"all":[{"value":0,"_account_id":16465,"name":"Kristi Nikolla","email":"knikolla@bu.edu","username":"knikolla"},{"value":2,"date":"2019-05-17 18:40:15.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"value":2,"date":"2019-05-12 13:35:47.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":0,"_account_id":2903,"name":"Morgan Fainberg","email":"morgan.fainberg@gmail.com","username":"mdrnstm"},{"value":0,"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"all":[{"value":0,"_account_id":16465,"name":"Kristi Nikolla","email":"knikolla@bu.edu","username":"knikolla"},{"value":1,"date":"2019-05-17 18:40:15.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"value":0,"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":0,"_account_id":2903,"name":"Morgan Fainberg","email":"morgan.fainberg@gmail.com","username":"mdrnstm"},{"value":0,"date":"2019-05-16 06:51:45.000000000","permitted_voting_range":{"min":0,"max":1},"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},{"_account_id":2903,"name":"Morgan Fainberg","email":"morgan.fainberg@gmail.com","username":"mdrnstm"},{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"_account_id":16465,"name":"Kristi Nikolla","email":"knikolla@bu.edu","username":"knikolla"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-05-08 18:31:24.000000000","updated_by":{"_account_id":16465,"name":"Kristi Nikolla","email":"knikolla@bu.edu","username":"knikolla"},"reviewer":{"_account_id":16465,"name":"Kristi Nikolla","email":"knikolla@bu.edu","username":"knikolla"},"state":"REVIEWER"},{"updated":"2019-05-12 13:35:47.000000000","updated_by":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"reviewer":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"state":"REVIEWER"},{"updated":"2019-05-16 06:51:45.000000000","updated_by":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"reviewer":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"state":"REVIEWER"},{"updated":"2019-05-17 18:40:15.000000000","updated_by":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"reviewer":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"state":"REVIEWER"},{"updated":"2019-05-17 18:48:48.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"e5b6beec4dd8e5aceaef469418522b64c58c4e14","author":{"_account_id":2903,"name":"Morgan Fainberg","email":"morgan.fainberg@gmail.com","username":"mdrnstm"},"date":"2019-05-08 18:29:46.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"4367efa7fc464649ba2946e3069c7700a6359213","author":{"_account_id":16465,"name":"Kristi Nikolla","email":"knikolla@bu.edu","username":"knikolla"},"date":"2019-05-08 18:31:24.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"b0f6dd5f217ae829446f445018625b711e858375","author":{"_account_id":2903,"name":"Morgan Fainberg","email":"morgan.fainberg@gmail.com","username":"mdrnstm"},"date":"2019-05-08 18:31:50.000000000","message":"Patch Set 1: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"982357f28d3cafc20c7505aa466470a98a523448","author":{"_account_id":2903,"name":"Morgan Fainberg","email":"morgan.fainberg@gmail.com","username":"mdrnstm"},"date":"2019-05-08 18:32:10.000000000","message":"Patch Set 2: Published edit on patch set 1.","accounts_in_message":[],"_revision_number":2},{"id":"cab22bd68070bc62894a16b6f271ca1741ed6b9b","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2019-05-08 18:33:44.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"3b0b18fa9313ea09e13499f3943ccaecc1aada1d","author":{"_account_id":2903,"name":"Morgan Fainberg","email":"morgan.fainberg@gmail.com","username":"mdrnstm"},"date":"2019-05-08 18:37:21.000000000","message":"Patch Set 2:\n\nConfirming that we want an empty repository. There is no history, there is currently no code yet. This is green field work. I expected the first patch to be be repo init content.","accounts_in_message":[],"_revision_number":2},{"id":"4e705ea15d9192f22753fbbd2ba6d2cb7dd4702a","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2019-05-08 18:38:01.000000000","message":"Patch Set 2:\n\nthanks for confirming","accounts_in_message":[],"_revision_number":2},{"id":"3573bde0ce7bcd2b14bf93fe02853aeca18ee265","author":{"_account_id":2903,"name":"Morgan Fainberg","email":"morgan.fainberg@gmail.com","username":"mdrnstm"},"date":"2019-05-08 18:38:17.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"6809da76c69b409adc433953b19c56fbb11a7b2a","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2019-05-08 18:38:34.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"52fa6b2dc5c5463b977250df70b1e668b9a26f9d","author":{"_account_id":2903,"name":"Morgan Fainberg","email":"morgan.fainberg@gmail.com","username":"mdrnstm"},"date":"2019-05-08 18:43:06.000000000","message":"Patch Set 2: Workflow-1\n\nHolding on merged, while I look at a couple related bits. I expect to unblock this, but want to avoid needless work for the infra team if something is wrong in an assumption I\u0027m making.","accounts_in_message":[],"_revision_number":2},{"id":"d1609d68b846b27ce3985a98eac02afbdb1602bf","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-08 18:44:16.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-zuul-jobs-linters http://logs.openstack.org/62/657862/2/check/openstack-zuul-jobs-linters/205d665/ : SUCCESS in 5m 52s\n- project-config-gerrit http://logs.openstack.org/62/657862/2/check/project-config-gerrit/06e04b7/ : SUCCESS in 3m 11s","accounts_in_message":[],"_revision_number":2},{"id":"5c507962195e36e14d496edaaaf2e159c58729c9","author":{"_account_id":2903,"name":"Morgan Fainberg","email":"morgan.fainberg@gmail.com","username":"mdrnstm"},"date":"2019-05-08 19:06:33.000000000","message":"Patch Set 2: -Workflow\n\nThis is fine as is looking into it.","accounts_in_message":[],"_revision_number":2},{"id":"3e044071d8d58e2c5d3a127deecf94921fd89a22","author":{"_account_id":2903,"name":"Morgan Fainberg","email":"morgan.fainberg@gmail.com","username":"mdrnstm"},"date":"2019-05-08 19:37:05.000000000","message":"Patch Set 3: Published edit on patch set 2.","accounts_in_message":[],"_revision_number":3},{"id":"ea7529268d816df6760ecb81a1dcb3c7c5c3958c","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-08 19:45:50.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-zuul-jobs-linters http://logs.openstack.org/62/657862/3/check/openstack-zuul-jobs-linters/e5febd9/ : SUCCESS in 5m 39s\n- project-config-gerrit http://logs.openstack.org/62/657862/3/check/project-config-gerrit/701efe4/ : SUCCESS in 3m 21s","accounts_in_message":[],"_revision_number":3},{"id":"59706ec465fab8044d6a9a02ad6b043ac96c5370","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2019-05-11 14:33:10.000000000","message":"Patch Set 3: Workflow-1\n\nSorry, we need to merge https://review.opendev.org/658123 first","accounts_in_message":[],"_revision_number":3},{"id":"7e2ded74aa5046ecb44f102fb462841ed96589e3","author":{"_account_id":2903,"name":"Morgan Fainberg","email":"morgan.fainberg@gmail.com","username":"mdrnstm"},"date":"2019-05-11 18:13:22.000000000","message":"Patch Set 3:\n\nNot a problem, merge this when the other one merges in.","accounts_in_message":[],"_revision_number":3},{"id":"2233fdd7e70e984251da6a77b6f100da1085c71b","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2019-05-12 13:35:03.000000000","message":"Patch Set 3: -Workflow","accounts_in_message":[],"_revision_number":3},{"id":"a47345c67c24c5de3e3dd6cd9fe320d1ac9f8b0b","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2019-05-12 13:35:47.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"732e45297e2216cb51b183d647bca28a86eca161","author":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"date":"2019-05-17 18:40:15.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"518db333da5d2cd817ef1436534ff44b0547b658","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-17 18:40:28.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"d80e0725adaf2aacb4b7c96b6dced0c7351e7412","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-17 18:48:48.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-zuul-jobs-linters http://logs.openstack.org/62/657862/3/gate/openstack-zuul-jobs-linters/0628912/ : SUCCESS in 6m 00s\n- project-config-gerrit http://logs.openstack.org/62/657862/3/gate/project-config-gerrit/6c3768a/ : SUCCESS in 3m 14s","accounts_in_message":[],"_revision_number":3},{"id":"be0c1eaabf79b4a423e7404c7204ac8dafbe38d5","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-17 18:48:48.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"7bd4fa8cbb7ab6f76f2ee16b0eede019329c4cf7","revisions":{"20e13da5fd41892400580d634e7925dae6096e65":{"kind":"REWORK","_number":1,"created":"2019-05-08 18:29:46.000000000","uploader":{"_account_id":2903,"name":"Morgan Fainberg","email":"morgan.fainberg@gmail.com","username":"mdrnstm"},"ref":"refs/changes/62/657862/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/62/657862/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/62/657862/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/62/657862/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/62/657862/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/62/657862/1"}}},"commit":{"parents":[{"commit":"6ed18827e769d27582a96f0de087324e89d2c0c6","subject":"Merge \"Create election repository for StarlingX\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/6ed18827e769d27582a96f0de087324e89d2c0c6"}]}],"author":{"name":"morgan fainberg","email":"morgan.fainberg@gmail.com","date":"2019-05-08 18:18:13.000000000","tz":-420},"committer":{"name":"morgan fainberg","email":"morgan.fainberg@gmail.com","date":"2019-05-08 18:27:47.000000000","tz":-420},"subject":"Add spectroscope repository","message":"Add spectroscope repository\n\nSpectroscope is intended to be an Identity Resource Broker Service.\nThis adds the initial spectroscope repository and ACL file.\n\nChange-Id: I1e67a913ac61ef78b9a57d564f3e50bbe2fe4a45\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/20e13da5fd41892400580d634e7925dae6096e65"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/20e13da5fd41892400580d634e7925dae6096e65"}]},"branch":"refs/heads/master"},"20be92af1523e39b21f4481c0858b9dd3552c9d2":{"kind":"REWORK","_number":2,"created":"2019-05-08 18:32:10.000000000","uploader":{"_account_id":2903,"name":"Morgan Fainberg","email":"morgan.fainberg@gmail.com","username":"mdrnstm"},"ref":"refs/changes/62/657862/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/62/657862/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/62/657862/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/62/657862/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/62/657862/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/62/657862/2"}}},"commit":{"parents":[{"commit":"6ed18827e769d27582a96f0de087324e89d2c0c6","subject":"Merge \"Create election repository for StarlingX\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/6ed18827e769d27582a96f0de087324e89d2c0c6"}]}],"author":{"name":"morgan fainberg","email":"morgan.fainberg@gmail.com","date":"2019-05-08 18:18:13.000000000","tz":-420},"committer":{"name":"Morgan Fainberg","email":"morgan.fainberg@gmail.com","date":"2019-05-08 18:32:08.000000000","tz":0},"subject":"Add spectroscope repository","message":"Add spectroscope repository\n\nSpectroscope is intended to be an Identity Resource Broker Service.\nThis adds the initial spectroscope repository and ACL file.\n\nChange-Id: I1e67a913ac61ef78b9a57d564f3e50bbe2fe4a45\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/20be92af1523e39b21f4481c0858b9dd3552c9d2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/20be92af1523e39b21f4481c0858b9dd3552c9d2"}]},"branch":"refs/heads/master"},"7bd4fa8cbb7ab6f76f2ee16b0eede019329c4cf7":{"kind":"REWORK","_number":3,"created":"2019-05-08 19:37:05.000000000","uploader":{"_account_id":2903,"name":"Morgan Fainberg","email":"morgan.fainberg@gmail.com","username":"mdrnstm"},"ref":"refs/changes/62/657862/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/62/657862/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/62/657862/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/62/657862/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/62/657862/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/62/657862/3"}}},"commit":{"parents":[{"commit":"6ed18827e769d27582a96f0de087324e89d2c0c6","subject":"Merge \"Create election repository for StarlingX\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/6ed18827e769d27582a96f0de087324e89d2c0c6"}]}],"author":{"name":"morgan fainberg","email":"morgan.fainberg@gmail.com","date":"2019-05-08 18:18:13.000000000","tz":-420},"committer":{"name":"Morgan Fainberg","email":"morgan.fainberg@gmail.com","date":"2019-05-08 19:37:03.000000000","tz":0},"subject":"Add spectroscope repository","message":"Add spectroscope repository\n\nSpectroscope is intended to be an Identity Resource Broker Service.\nThis adds the initial spectroscope repository and ACL file.\n\nChange-Id: I1e67a913ac61ef78b9a57d564f3e50bbe2fe4a45\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/7bd4fa8cbb7ab6f76f2ee16b0eede019329c4cf7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/7bd4fa8cbb7ab6f76f2ee16b0eede019329c4cf7"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
