)]}'
{"id":"openstack%2Fproject-config~658179","triplet_id":"openstack%2Fproject-config~master~I418073c7cb6016bf2a10947815816642b0d38662","project":"openstack/project-config","branch":"master","topic":"remove-vault-config","hashtags":[],"change_id":"I418073c7cb6016bf2a10947815816642b0d38662","subject":"Remove charm-vault from zuul.d/projects.yaml","status":"MERGED","created":"2019-05-09 18:41:17.000000000","updated":"2019-05-13 20:58:42.000000000","submitted":"2019-05-13 20:58:42.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"658179-1557781122341-b2ea1c39","meta_rev_id":"1b6e537a8e63fc957c36dbc2c15ca65893f1e1f6","_number":658179,"virtual_id_number":658179,"owner":{"_account_id":26040,"name":"Pete Vander Giessen","email":"pete.vandergiessen@canonical.com","username":"petevg"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":30418,"name":"George Kraft","email":"george.kraft@canonical.com","username":"cynerva"},{"value":0,"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"value":0,"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":0,"_account_id":20805,"name":"David Ames","email":"david@davam.net","username":"thedac"},{"value":0,"_account_id":935,"name":"James Page","email":"ringo.page@gmail.com","username":"james-page"},{"value":0,"date":"2019-05-09 19:13:23.000000000","_account_id":26040,"name":"Pete Vander Giessen","email":"pete.vandergiessen@canonical.com","username":"petevg"},{"value":2,"date":"2019-05-13 20:58:41.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"recommended":{"_account_id":30418,"name":"George Kraft","email":"george.kraft@canonical.com","username":"cynerva"},"all":[{"value":1,"date":"2019-05-09 21:25:41.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":30418,"name":"George Kraft","email":"george.kraft@canonical.com","username":"cynerva"},{"value":2,"date":"2019-05-13 20:46:19.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"value":2,"date":"2019-05-10 06:16:00.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":1,"date":"2019-05-09 20:01:25.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":20805,"name":"David Ames","email":"david@davam.net","username":"thedac"},{"value":1,"date":"2019-05-10 14:06:48.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":935,"name":"James Page","email":"ringo.page@gmail.com","username":"james-page"},{"value":0,"_account_id":26040,"name":"Pete Vander Giessen","email":"pete.vandergiessen@canonical.com","username":"petevg"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"all":[{"value":0,"_account_id":30418,"name":"George Kraft","email":"george.kraft@canonical.com","username":"cynerva"},{"value":1,"date":"2019-05-13 20:46:19.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"value":0,"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":0,"_account_id":20805,"name":"David Ames","email":"david@davam.net","username":"thedac"},{"value":0,"_account_id":935,"name":"James Page","email":"ringo.page@gmail.com","username":"james-page"},{"value":0,"_account_id":26040,"name":"Pete Vander Giessen","email":"pete.vandergiessen@canonical.com","username":"petevg"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":935,"name":"James Page","email":"ringo.page@gmail.com","username":"james-page"},{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"_account_id":20805,"name":"David Ames","email":"david@davam.net","username":"thedac"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":26040,"name":"Pete Vander Giessen","email":"pete.vandergiessen@canonical.com","username":"petevg"},{"_account_id":30418,"name":"George Kraft","email":"george.kraft@canonical.com","username":"cynerva"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-05-09 20:01:25.000000000","updated_by":{"_account_id":20805,"name":"David Ames","email":"david@davam.net","username":"thedac"},"reviewer":{"_account_id":20805,"name":"David Ames","email":"david@davam.net","username":"thedac"},"state":"REVIEWER"},{"updated":"2019-05-09 21:25:41.000000000","updated_by":{"_account_id":30418,"name":"George Kraft","email":"george.kraft@canonical.com","username":"cynerva"},"reviewer":{"_account_id":30418,"name":"George Kraft","email":"george.kraft@canonical.com","username":"cynerva"},"state":"REVIEWER"},{"updated":"2019-05-10 06:16:00.000000000","updated_by":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"reviewer":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"state":"REVIEWER"},{"updated":"2019-05-10 14:06:48.000000000","updated_by":{"_account_id":935,"name":"James Page","email":"ringo.page@gmail.com","username":"james-page"},"reviewer":{"_account_id":935,"name":"James Page","email":"ringo.page@gmail.com","username":"james-page"},"state":"REVIEWER"},{"updated":"2019-05-13 20:46:19.000000000","updated_by":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"reviewer":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"state":"REVIEWER"},{"updated":"2019-05-13 20:58:41.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"4a60ddeb7c91466285b749917ea15dc9a4d5d23d","author":{"_account_id":26040,"name":"Pete Vander Giessen","email":"pete.vandergiessen@canonical.com","username":"petevg"},"date":"2019-05-09 18:41:17.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"3487daaebf8ea421a2e5295f5c89a9c25d120bb3","author":{"_account_id":26040,"name":"Pete Vander Giessen","email":"pete.vandergiessen@canonical.com","username":"petevg"},"date":"2019-05-09 18:43:52.000000000","message":"Uploaded patch set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"84bf23aaa41e203d378538976bd80bc57cd33386","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2019-05-09 18:52:59.000000000","message":"Patch Set 2:\n\nLooking at https://opendev.org/x/charm-vault/src/branch/master/.zuul.yaml you will not be able to merge any changes to x/charm-vault once this merges *unless* you add a \"noop\" job to the gate queue.\n\nThe cover-jobs template has no gate jobs and you always need a gate job to merge anything...\n\nAlso, since you\u0027re not a core of charm-vault, I\u0027d like to see some cores +1 this change.","accounts_in_message":[],"_revision_number":2},{"id":"bd00a5a50df0921e0a6459618ad726d7d541f482","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2019-05-09 18:53:41.000000000","message":"Patch Set 2:\n\nsorry, see you are a core - still, please ask some other cores to chime in...","accounts_in_message":[],"_revision_number":2},{"id":"af0fae05d48ed3063161e81e62863c30803fabb4","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-09 18:55:23.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-zuul-jobs-linters http://logs.openstack.org/79/658179/2/check/openstack-zuul-jobs-linters/f783886/ : SUCCESS in 6m 06s","accounts_in_message":[],"_revision_number":2},{"id":"58c1a042eb71461a20dbe2895ab9c4818a3f8ea3","author":{"_account_id":20805,"name":"David Ames","email":"david@davam.net","username":"thedac"},"date":"2019-05-09 19:01:38.000000000","message":"Patch Set 2: Code-Review+1\n\nLooks good.","accounts_in_message":[],"_revision_number":2},{"id":"ccab13d52a0e561acf294c28a171bc866574c241","author":{"_account_id":20805,"name":"David Ames","email":"david@davam.net","username":"thedac"},"date":"2019-05-09 19:02:20.000000000","message":"Patch Set 2: -Code-Review\n\nWhoops, missed Andreas comment.","accounts_in_message":[],"_revision_number":2},{"id":"2dc239f9bc307f088c5a15c6377e686d89f02080","author":{"_account_id":26040,"name":"Pete Vander Giessen","email":"pete.vandergiessen@canonical.com","username":"petevg"},"date":"2019-05-09 19:07:57.000000000","message":"Patch Set 2:\n\n@Andreas Jaeger: The blocked merge proposal adds python35-charm-jobs to the .zuul.yaml. https://review.opendev.org/#/c/658153/3\n\nIs that sufficient to address the concern that you bring up? Or does the change to the .zuul.yaml need to get merged before we make this change (and thus before the other changes in that mp are merged)?","accounts_in_message":[],"_revision_number":2},{"id":"29fd2a98e0fc6486ecb2e64da47b23a5a1670a8a","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2019-05-09 19:11:40.000000000","message":"Patch Set 2:\n\nYes, merging a change that adds python35 jobs will be enough...\n\nso, we can merge this - and then you need to merge a change that adds job for the gate queue like python35 jobs...\n\nIf that\u0027s how the charm-vault team want to move forward, ok with me - but please ask for some +1s..","accounts_in_message":[],"_revision_number":2},{"id":"5f3ef4da6dece6ab7fdefaa6c996839f1ef2b462","author":{"_account_id":26040,"name":"Pete Vander Giessen","email":"pete.vandergiessen@canonical.com","username":"petevg"},"date":"2019-05-09 19:13:23.000000000","message":"Patch Set 2:\n\n@Andreas Jaeger: thx for the confirmation! +1s on this end I think are just waiting on a lunch break :-)","accounts_in_message":[],"_revision_number":2},{"id":"44039a0c1e810cddeaa20354e8b70166f66f18ba","author":{"_account_id":20805,"name":"David Ames","email":"david@davam.net","username":"thedac"},"date":"2019-05-09 20:01:25.000000000","message":"Patch Set 2: Code-Review+1\n\nOK, looks like we have it sorted. Confirmed we will be adding the .zuul.yaml change immediately after this lands.","accounts_in_message":[],"_revision_number":2},{"id":"ddfe6d1b4c441bed1203f7fab0d71ba40f1f8d68","author":{"_account_id":30418,"name":"George Kraft","email":"george.kraft@canonical.com","username":"cynerva"},"date":"2019-05-09 21:25:41.000000000","message":"Patch Set 2: Code-Review+1\n\nLooks good to me.","accounts_in_message":[],"_revision_number":2},{"id":"e2ae9a238c52065470a1dca60f12f19f0fbebdc8","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2019-05-10 06:16:00.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"5dc92c535b02b96c388e192b66a6dc58a032c9fa","author":{"_account_id":935,"name":"James Page","email":"ringo.page@gmail.com","username":"james-page"},"date":"2019-05-10 14:06:48.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"9ceb039cef96fe079823bf6fe80b34f02751d203","author":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"date":"2019-05-13 20:46:19.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"bfb3d9acf68f4e9a72ba87e6f54c303eca5cba59","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-13 20:46:52.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"0fbee7d4d98fdf8fa32a73a06859bc2270abf750","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-13 20:58:41.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-zuul-jobs-linters http://logs.openstack.org/79/658179/2/gate/openstack-zuul-jobs-linters/2a5aae7/ : SUCCESS in 5m 29s","accounts_in_message":[],"_revision_number":2},{"id":"f9d1ccc0a0c1ae9453f6d2632924595fdce9551a","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-13 20:58:42.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"074899f2505ed89b3ebc6fa9e7457f4c567a1142","revisions":{"ca6cc0d9fb9c3317a6dbab117c6a18503395d1b4":{"kind":"REWORK","_number":1,"created":"2019-05-09 18:41:17.000000000","uploader":{"_account_id":26040,"name":"Pete Vander Giessen","email":"pete.vandergiessen@canonical.com","username":"petevg"},"ref":"refs/changes/79/658179/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/79/658179/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/79/658179/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/79/658179/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/79/658179/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/79/658179/1"}}},"commit":{"parents":[{"commit":"c43dc3f1ed38b5c239c2d682cca652a2b62a656c","subject":"Merge \"Add testinfra 3rd party testing\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/c43dc3f1ed38b5c239c2d682cca652a2b62a656c"}]}],"author":{"name":"Pete Vander Giessen","email":"pete.vandergiessen@canonical.com","date":"2019-05-09 18:40:17.000000000","tz":-240},"committer":{"name":"Pete Vander Giessen","email":"pete.vandergiessen@canonical.com","date":"2019-05-09 18:41:16.000000000","tz":-240},"subject":"Remove charm-vault from zuul.d/projects.yaml","message":"Remove charm-vault from zuul.d/projects.yaml\n\ncharm-vaul has a .zuul.yaml to govern the tests that it runs. There is\nno need to clobber with the defaults here.\n\nThis addresses an issue where a merge proposal is blocked by\nerroneously failing python 2.7 tests.\n\nChange-Id: I418073c7cb6016bf2a10947815816642b0d38662\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/ca6cc0d9fb9c3317a6dbab117c6a18503395d1b4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/ca6cc0d9fb9c3317a6dbab117c6a18503395d1b4"}]},"branch":"refs/heads/master"},"074899f2505ed89b3ebc6fa9e7457f4c567a1142":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2019-05-09 18:43:52.000000000","uploader":{"_account_id":26040,"name":"Pete Vander Giessen","email":"pete.vandergiessen@canonical.com","username":"petevg"},"ref":"refs/changes/79/658179/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/79/658179/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/79/658179/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/79/658179/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/79/658179/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/79/658179/2"}}},"commit":{"parents":[{"commit":"c43dc3f1ed38b5c239c2d682cca652a2b62a656c","subject":"Merge \"Add testinfra 3rd party testing\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/c43dc3f1ed38b5c239c2d682cca652a2b62a656c"}]}],"author":{"name":"Pete Vander Giessen","email":"pete.vandergiessen@canonical.com","date":"2019-05-09 18:40:17.000000000","tz":-240},"committer":{"name":"Pete Vander Giessen","email":"pete.vandergiessen@canonical.com","date":"2019-05-09 18:43:37.000000000","tz":-240},"subject":"Remove charm-vault from zuul.d/projects.yaml","message":"Remove charm-vault from zuul.d/projects.yaml\n\ncharm-vault has a .zuul.yaml to govern the tests that it runs. There\nis no need to clobber with the defaults here.\n\nThis addresses an issue where a merge proposal is blocked by\nerroneously failing python 2.7 tests.\n\nChange-Id: I418073c7cb6016bf2a10947815816642b0d38662\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/074899f2505ed89b3ebc6fa9e7457f4c567a1142"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/074899f2505ed89b3ebc6fa9e7457f4c567a1142"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
