)]}'
{"id":"openstack%2Fproject-config~659620","triplet_id":"openstack%2Fproject-config~master~I485a09759124258dc24ad4be9f0e25c0e216c1e1","project":"openstack/project-config","branch":"master","topic":"zuul-nodejs10","hashtags":[],"change_id":"I485a09759124258dc24ad4be9f0e25c0e216c1e1","subject":"Update node to version 10 for release-zuul-python","status":"MERGED","created":"2019-05-16 16:40:11.000000000","updated":"2019-05-16 17:20:39.000000000","submitted":"2019-05-16 17:20:38.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"659620-1558027239160-ec554898","meta_rev_id":"cedd84a641cedd1f0c4538d7494cc183e99f6bfd","_number":659620,"virtual_id_number":659620,"owner":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"value":0,"_account_id":4162,"name":"Paul Belanger","email":"pabelanger@redhat.com","username":"pabelanger"},{"value":0,"_account_id":16068,"name":"Tobias Henkel","email":"tobias.henkel@bmw.de","username":"tobias.henkel"},{"value":2,"date":"2019-05-16 17:20:38.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"all":[{"value":2,"date":"2019-05-16 16:51:23.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"value":2,"date":"2019-05-16 16:52:40.000000000","_account_id":4162,"name":"Paul Belanger","email":"pabelanger@redhat.com","username":"pabelanger"},{"value":0,"_account_id":16068,"name":"Tobias Henkel","email":"tobias.henkel@bmw.de","username":"tobias.henkel"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":4162,"name":"Paul Belanger","email":"pabelanger@redhat.com","username":"pabelanger"},"all":[{"value":0,"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"value":1,"date":"2019-05-16 16:52:40.000000000","_account_id":4162,"name":"Paul Belanger","email":"pabelanger@redhat.com","username":"pabelanger"},{"value":0,"_account_id":16068,"name":"Tobias Henkel","email":"tobias.henkel@bmw.de","username":"tobias.henkel"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"_account_id":4162,"name":"Paul Belanger","email":"pabelanger@redhat.com","username":"pabelanger"},{"_account_id":16068,"name":"Tobias Henkel","email":"tobias.henkel@bmw.de","username":"tobias.henkel"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-05-16 16:40:55.000000000","updated_by":{"_account_id":16068,"name":"Tobias Henkel","email":"tobias.henkel@bmw.de","username":"tobias.henkel"},"reviewer":{"_account_id":16068,"name":"Tobias Henkel","email":"tobias.henkel@bmw.de","username":"tobias.henkel"},"state":"REVIEWER"},{"updated":"2019-05-16 16:51:23.000000000","updated_by":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"reviewer":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"state":"REVIEWER"},{"updated":"2019-05-16 16:52:40.000000000","updated_by":{"_account_id":4162,"name":"Paul Belanger","email":"pabelanger@redhat.com","username":"pabelanger"},"reviewer":{"_account_id":4162,"name":"Paul Belanger","email":"pabelanger@redhat.com","username":"pabelanger"},"state":"REVIEWER"},{"updated":"2019-05-16 17:20:38.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"c415a9d11ab79a8c71accc442a30f357e2d067c7","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2019-05-16 16:40:11.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"28284f5941cc1a81552d69d5a367a2a2b05be27e","author":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"date":"2019-05-16 16:40:47.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"4580b51368804fcb9d0aca56f280cc14bb6a33b6","author":{"_account_id":16068,"name":"Tobias Henkel","email":"tobias.henkel@bmw.de","username":"tobias.henkel"},"date":"2019-05-16 16:40:55.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"9c63fcf6969b2d83fca16252adc0f6fc64671ef3","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-16 16:42:58.000000000","message":"Patch Set 1: Verified-1\n\n(1 comment)\n\nZuul encountered a syntax error while parsing its configuration in the\nrepo openstack/project-config on branch master.  The error was:\n\n  expected dict for dictionary value @ data[\u0027vars\u0027]\n\nThe error appears in the following job stanza:\n\n  job:\n      name: release-zuul-python\n      parent: publish-openstack-artifacts\n      description: |\n        Release python tarballs / wheels to pypi for zuul projects.\n      pre-run: playbooks/zuul-tarball/pre.yaml\n      run: playbooks/zuul-tarball/run.yaml\n      post-run:\n        - playbooks/zuul-tarball/post.yaml\n        - playbooks/publish/pypi.yaml\n      secrets:\n        - secret: pypi_secret\n          name: pypi_info\n        - gpg_key\n      vars:\n        - node_version: 10\n\n  in \"openstack/project-config/zuul.d/jobs.yaml@master\", line 156, column 3","accounts_in_message":[],"_revision_number":1},{"id":"989aa45ab81ef2cf21e9ffdfed1d2f0b6cdfe143","author":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"date":"2019-05-16 16:51:08.000000000","message":"Patch Set 2: Published edit on patch set 1.","accounts_in_message":[],"_revision_number":2},{"id":"0eae0cea261ff236b9063ae36b747363b78896f0","author":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"date":"2019-05-16 16:51:23.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"92f23448b94805f569c927ddd174c7ae37d76ee5","author":{"_account_id":4162,"name":"Paul Belanger","email":"pabelanger@redhat.com","username":"pabelanger"},"date":"2019-05-16 16:52:40.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"c58ef5bcc229faa2ded7b2c639862b543ecb7fe5","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-16 17:07:02.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-zuul-jobs-linters http://logs.openstack.org/20/659620/2/check/openstack-zuul-jobs-linters/1bfa64a/ : SUCCESS in 6m 28s","accounts_in_message":[],"_revision_number":2},{"id":"418d3d5c09b356b6787ec8f19bec93bfd1138fc1","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-16 17:07:19.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"d8d07eb410caf37d84c3d2edcfbea723803839f9","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-16 17:20:38.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-zuul-jobs-linters http://logs.openstack.org/20/659620/2/gate/openstack-zuul-jobs-linters/26d9dee/ : SUCCESS in 6m 51s","accounts_in_message":[],"_revision_number":2},{"id":"3add4628258c5c260baf55e4b559f6c627c5a5da","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-16 17:20:39.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"657191ae3ef37936712e01235ad137f95f5f681c","revisions":{"e0d01b4ae7b08344f2e8a39eb4529f6152f16ca0":{"kind":"REWORK","_number":1,"created":"2019-05-16 16:40:11.000000000","uploader":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"ref":"refs/changes/20/659620/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/20/659620/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/20/659620/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/20/659620/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/20/659620/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/20/659620/1"}}},"commit":{"parents":[{"commit":"2341ca83f21ebfa5c594c1989ac922dcb5e2e58b","subject":"Merge \"Set up cyborg-tempest-plugin repository\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/2341ca83f21ebfa5c594c1989ac922dcb5e2e58b"}]}],"author":{"name":"Andreas Jaeger","email":"aj@suse.com","date":"2019-05-16 16:39:14.000000000","tz":120},"committer":{"name":"Andreas Jaeger","email":"aj@suse.com","date":"2019-05-16 16:39:14.000000000","tz":120},"subject":"Update node to version 10 for release-zuul-python","message":"Update node to version 10 for release-zuul-python\n\nFollow https://opendev.org/zuul/zuul/src/branch/master/.zuul.yaml#L221\nand use nodejs 10.\n\nThis should fix the job, it currently has a too low yarn version\nfor the rest of the repo.\n\nChange-Id: I485a09759124258dc24ad4be9f0e25c0e216c1e1\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/e0d01b4ae7b08344f2e8a39eb4529f6152f16ca0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/e0d01b4ae7b08344f2e8a39eb4529f6152f16ca0"}]},"branch":"refs/heads/master"},"657191ae3ef37936712e01235ad137f95f5f681c":{"kind":"REWORK","_number":2,"created":"2019-05-16 16:51:08.000000000","uploader":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"ref":"refs/changes/20/659620/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/20/659620/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/20/659620/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/20/659620/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/20/659620/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/20/659620/2"}}},"commit":{"parents":[{"commit":"2341ca83f21ebfa5c594c1989ac922dcb5e2e58b","subject":"Merge \"Set up cyborg-tempest-plugin repository\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/2341ca83f21ebfa5c594c1989ac922dcb5e2e58b"}]}],"author":{"name":"Andreas Jaeger","email":"aj@suse.com","date":"2019-05-16 16:39:14.000000000","tz":120},"committer":{"name":"Clark Boylan","email":"cboylan@sapwetik.org","date":"2019-05-16 16:51:06.000000000","tz":0},"subject":"Update node to version 10 for release-zuul-python","message":"Update node to version 10 for release-zuul-python\n\nFollow https://opendev.org/zuul/zuul/src/branch/master/.zuul.yaml#L221\nand use nodejs 10.\n\nThis should fix the job, it currently has a too low yarn version\nfor the rest of the repo.\n\nChange-Id: I485a09759124258dc24ad4be9f0e25c0e216c1e1\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/657191ae3ef37936712e01235ad137f95f5f681c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/657191ae3ef37936712e01235ad137f95f5f681c"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
