)]}'
{"id":"openstack%2Fproject-config~659810","triplet_id":"openstack%2Fproject-config~master~Ifdbd75272e525b499fe97cbf72e9d735fd13876e","project":"openstack/project-config","branch":"master","topic":"requirements","hashtags":[],"change_id":"Ifdbd75272e525b499fe97cbf72e9d735fd13876e","subject":"Update to ansible 2.7 for linting","status":"MERGED","created":"2019-05-17 14:36:21.000000000","updated":"2019-06-02 16:15:48.000000000","submitted":"2019-06-02 16:15:48.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"659810-1559492148947-02bd17ee","meta_rev_id":"fa870a9399c48a41b7217bc7743c49f93d54c197","_number":659810,"virtual_id_number":659810,"owner":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":4162,"name":"Paul Belanger","email":"pabelanger@redhat.com","username":"pabelanger"},{"value":0,"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"value":0,"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":0,"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},{"value":2,"date":"2019-06-02 16:15:48.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":4162,"name":"Paul Belanger","email":"pabelanger@redhat.com","username":"pabelanger"},"all":[{"value":2,"date":"2019-05-17 14:51:13.000000000","_account_id":4162,"name":"Paul Belanger","email":"pabelanger@redhat.com","username":"pabelanger"},{"value":2,"date":"2019-05-23 21:32:20.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"value":0,"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":2,"date":"2019-06-02 16:04:54.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"all":[{"value":0,"_account_id":4162,"name":"Paul Belanger","email":"pabelanger@redhat.com","username":"pabelanger"},{"value":0,"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"value":0,"date":"2019-05-24 04:53:21.000000000","permitted_voting_range":{"min":0,"max":1},"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":1,"date":"2019-06-02 16:04:54.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"_account_id":4162,"name":"Paul Belanger","email":"pabelanger@redhat.com","username":"pabelanger"},{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-05-17 14:51:13.000000000","updated_by":{"_account_id":4162,"name":"Paul Belanger","email":"pabelanger@redhat.com","username":"pabelanger"},"reviewer":{"_account_id":4162,"name":"Paul Belanger","email":"pabelanger@redhat.com","username":"pabelanger"},"state":"REVIEWER"},{"updated":"2019-05-23 21:32:20.000000000","updated_by":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"reviewer":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"state":"REVIEWER"},{"updated":"2019-06-02 16:04:54.000000000","updated_by":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"reviewer":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"state":"REVIEWER"},{"updated":"2019-06-02 16:15:48.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"29f3a2e9fb25adc5c062db10ee1abdfe02b518a9","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2019-05-17 14:36:21.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"6a3cb2919331aee5d912ef88c752cd81d47adf0a","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2019-05-17 14:37:40.000000000","message":"Topic changed from update-hacking to requirements","accounts_in_message":[],"_revision_number":1},{"id":"3aee8bc500e49e2387c44b5e53ba6c5b564a84a4","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-17 14:47:44.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- build-openstack-specs-site http://logs.openstack.org/10/659810/1/check/build-openstack-specs-site/9a551ab/ : SUCCESS in 2m 53s\n- openstack-zuul-jobs-linters http://logs.openstack.org/10/659810/1/check/openstack-zuul-jobs-linters/ccfa81d/ : SUCCESS in 5m 51s\n- project-config-dib http://logs.openstack.org/10/659810/1/check/project-config-dib/f1733c4/ : SUCCESS in 2m 32s\n- project-config-gerrit http://logs.openstack.org/10/659810/1/check/project-config-gerrit/83d2ab2/ : SUCCESS in 3m 03s\n- project-config-grafyaml http://logs.openstack.org/10/659810/1/check/project-config-grafyaml/fe246d8/ : SUCCESS in 2m 49s\n- project-config-infra-docs-index http://logs.openstack.org/10/659810/1/check/project-config-infra-docs-index/5fe52a4/ : SUCCESS in 2m 12s\n- project-config-irc-access http://logs.openstack.org/10/659810/1/check/project-config-irc-access/b163bbf/ : SUCCESS in 6m 20s\n- project-config-nodepool http://logs.openstack.org/10/659810/1/check/project-config-nodepool/c09b0b7/ : SUCCESS in 3m 26s","accounts_in_message":[],"_revision_number":1},{"id":"e80f75d322805e0a1dacf1d486760ad2cef51c99","author":{"_account_id":4162,"name":"Paul Belanger","email":"pabelanger@redhat.com","username":"pabelanger"},"date":"2019-05-17 14:51:13.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"a89d88bfc71daa2fee489a49c6b8ec08de19c0bd","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2019-05-17 14:58:51.000000000","message":"Patch Set 1: Workflow-1\n\nLet\u0027s discuss on zuul-discuss list a bit first.","accounts_in_message":[],"_revision_number":1},{"id":"2e80c12080f64c3d95ec0d4540185aceacfd3113","author":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"date":"2019-05-23 21:32:20.000000000","message":"Patch Set 1: Code-Review+2\n\nI think we can make this change safely.","accounts_in_message":[],"_revision_number":1},{"id":"c0d2a4985bd87d5d494e39118223653102b5f3b6","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2019-05-24 04:53:21.000000000","message":"Patch Set 1: -Workflow","accounts_in_message":[],"_revision_number":1},{"id":"6d170b83b0fcbbaef36551f52ccc37542f0d0fc7","author":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"date":"2019-06-02 16:04:54.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"1cff712f69794f578f4032f08e739bbd8cb8b1c1","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-02 16:05:12.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"098a7a8d8132b5ce2ab31dfbae18de34a45057d3","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-02 16:15:48.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"fa870a9399c48a41b7217bc7743c49f93d54c197","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-02 16:15:48.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- build-openstack-specs-site http://logs.openstack.org/10/659810/1/gate/build-openstack-specs-site/dda42b3/ : SUCCESS in 3m 18s\n- openstack-zuul-jobs-linters http://logs.openstack.org/10/659810/1/gate/openstack-zuul-jobs-linters/bfef6a3/ : SUCCESS in 6m 16s\n- project-config-dib http://logs.openstack.org/10/659810/1/gate/project-config-dib/3befe0d/ : SUCCESS in 3m 37s\n- project-config-gerrit http://logs.openstack.org/10/659810/1/gate/project-config-gerrit/5b123a6/ : SUCCESS in 3m 30s\n- project-config-grafyaml http://logs.openstack.org/10/659810/1/gate/project-config-grafyaml/091ee72/ : SUCCESS in 3m 02s\n- project-config-infra-docs-index http://logs.openstack.org/10/659810/1/gate/project-config-infra-docs-index/83f7242/ : SUCCESS in 2m 56s\n- project-config-irc-access http://logs.openstack.org/10/659810/1/gate/project-config-irc-access/16dc713/ : SUCCESS in 6m 49s\n- project-config-nodepool http://logs.openstack.org/10/659810/1/gate/project-config-nodepool/2867241/ : SUCCESS in 4m 15s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"5e11aa650c7b0da03ec9ac3f3b69d2c466ca9834","revisions":{"5e11aa650c7b0da03ec9ac3f3b69d2c466ca9834":{"kind":"REWORK","_number":1,"created":"2019-05-17 14:36:21.000000000","uploader":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"ref":"refs/changes/10/659810/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/10/659810/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/10/659810/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/10/659810/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/10/659810/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/10/659810/1"}}},"commit":{"parents":[{"commit":"ecc2997769a6606a42acb76ddcc21b9ee4d7016f","subject":"Update hacking to version 1.1","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/ecc2997769a6606a42acb76ddcc21b9ee4d7016f"}]}],"author":{"name":"Andreas Jaeger","email":"aj@suse.com","date":"2019-05-17 14:35:43.000000000","tz":120},"committer":{"name":"Andreas Jaeger","email":"aj@suse.com","date":"2019-05-17 14:35:43.000000000","tz":120},"subject":"Update to ansible 2.7 for linting","message":"Update to ansible 2.7 for linting\n\nWe run ansible 2.7 with zuul, so update the version we use for linting.\n\nChange-Id: Ifdbd75272e525b499fe97cbf72e9d735fd13876e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/5e11aa650c7b0da03ec9ac3f3b69d2c466ca9834"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/5e11aa650c7b0da03ec9ac3f3b69d2c466ca9834"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
