)]}'
{"id":"openstack%2Fproject-config~660388","triplet_id":"openstack%2Fproject-config~master~I7c673f5a67a5fcd3839b4e210db44a1363e14fe3","project":"openstack/project-config","branch":"master","topic":"infra-base-images","hashtags":[],"change_id":"I7c673f5a67a5fcd3839b4e210db44a1363e14fe3","subject":"Add control plane mirror-only cloud regions","status":"ABANDONED","created":"2019-05-21 13:30:40.000000000","updated":"2022-02-17 13:37:14.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"f9a288f7b2cc4613b803a5c2982171f59c7383b1","_number":660388,"virtual_id_number":660388,"owner":{"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":1,"date":"2019-05-21 18:06:59.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-05-21 18:06:59.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"527834bd1cf6329992c457c8aa841711e8eef383","author":{"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},"date":"2019-05-21 13:30:40.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"8a6ac4a41c2f331a3f183673325c436ad7491f7a","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-21 13:45:15.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-zuul-jobs-linters http://logs.openstack.org/88/660388/1/check/openstack-zuul-jobs-linters/84a50ba/ : SUCCESS in 7m 30s\n- project-config-nodepool http://logs.openstack.org/88/660388/1/check/project-config-nodepool/2c4e065/ : SUCCESS in 4m 49s","accounts_in_message":[],"_revision_number":1},{"id":"68a888d1fcc24de2466cb3ff5873cecb6ca1e42b","author":{"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},"date":"2019-05-21 17:49:47.000000000","message":"Uploaded patch set 2: Patch Set 1 was rebased.","accounts_in_message":[],"_revision_number":2},{"id":"16a514033365f280cda69119333944f502331ccf","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-21 18:06:59.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-zuul-jobs-linters http://logs.openstack.org/88/660388/2/check/openstack-zuul-jobs-linters/3e73683/ : SUCCESS in 5m 34s\n- project-config-nodepool http://logs.openstack.org/88/660388/2/check/project-config-nodepool/87bfa54/ : SUCCESS in 3m 19s","accounts_in_message":[],"_revision_number":2},{"id":"f9a288f7b2cc4613b803a5c2982171f59c7383b1","tag":"autogenerated:gerrit:abandon","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2022-02-17 13:37:14.000000000","message":"Abandoned\n\nThis was essentially obsoleted in I610dae5d6d7fe01e15f66970b59d3b999bebb766","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"e105fab6ef607ba56c169859c6a8ba0119850fe9","revisions":{"2c3c315c673420ba9b85d4583aab30a063b8e0cf":{"kind":"REWORK","_number":1,"created":"2019-05-21 13:30:40.000000000","uploader":{"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},"ref":"refs/changes/88/660388/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/88/660388/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/88/660388/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/88/660388/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/88/660388/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/88/660388/1"}}},"commit":{"parents":[{"commit":"737cfd9fac16795b3b3c232719102e9fef022bb4","subject":"Build ubuntu bionic minimal images for control plane","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/737cfd9fac16795b3b3c232719102e9fef022bb4"}]}],"author":{"name":"Monty Taylor","email":"mordred@inaugust.com","date":"2019-05-21 13:28:14.000000000","tz":-300},"committer":{"name":"Monty Taylor","email":"mordred@inaugust.com","date":"2019-05-21 13:29:28.000000000","tz":-300},"subject":"Add control plane mirror-only cloud regions","message":"Add control plane mirror-only cloud regions\n\nSince we\u0027re building control plane base images, we might as well\ndo so for our mirror projects so that we can stay consistent.\n\nDepends-On https://review.opendev.org/660384\nChange-Id: I7c673f5a67a5fcd3839b4e210db44a1363e14fe3\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/2c3c315c673420ba9b85d4583aab30a063b8e0cf"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/2c3c315c673420ba9b85d4583aab30a063b8e0cf"}]},"branch":"refs/heads/master"},"e105fab6ef607ba56c169859c6a8ba0119850fe9":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2019-05-21 17:49:47.000000000","uploader":{"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},"ref":"refs/changes/88/660388/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/88/660388/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/88/660388/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/88/660388/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/88/660388/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/88/660388/2"}}},"commit":{"parents":[{"commit":"0c3e41384a0582e86a9fd90ec316e760d3d3152f","subject":"Build ubuntu bionic minimal images for control plane","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/0c3e41384a0582e86a9fd90ec316e760d3d3152f"}]}],"author":{"name":"Monty Taylor","email":"mordred@inaugust.com","date":"2019-05-21 13:28:14.000000000","tz":-300},"committer":{"name":"Monty Taylor","email":"mordred@inaugust.com","date":"2019-05-21 17:49:42.000000000","tz":-300},"subject":"Add control plane mirror-only cloud regions","message":"Add control plane mirror-only cloud regions\n\nSince we\u0027re building control plane base images, we might as well\ndo so for our mirror projects so that we can stay consistent.\n\nDepends-On https://review.opendev.org/660384\nChange-Id: I7c673f5a67a5fcd3839b4e210db44a1363e14fe3\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/e105fab6ef607ba56c169859c6a8ba0119850fe9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/e105fab6ef607ba56c169859c6a8ba0119850fe9"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
