)]}'
{"id":"openstack%2Fproject-config~660454","triplet_id":"openstack%2Fproject-config~master~I16edd504036a1bc0f9510db50b94a4eaa0dc7518","project":"openstack/project-config","branch":"master","topic":"sheepdog","hashtags":[],"change_id":"I16edd504036a1bc0f9510db50b94a4eaa0dc7518","subject":"Remove devstack-plugin-sheepdog","status":"MERGED","created":"2019-05-21 16:49:09.000000000","updated":"2019-12-26 10:19:29.000000000","submitted":"2019-12-26 10:19:29.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"660454-1577355569042-71032318","meta_rev_id":"e116ad6ac1c106bb84b58c28bd82680bba6a11d1","_number":660454,"virtual_id_number":660454,"owner":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"date":"2019-05-21 20:08:58.000000000","_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"value":0,"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2019-12-26 10:19:28.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":2,"date":"2019-05-21 16:52:35.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"value":2,"date":"2019-12-26 10:12:53.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"value":1,"date":"2019-12-26 10:12:53.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-05-21 16:52:35.000000000","updated_by":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"reviewer":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"state":"REVIEWER"},{"updated":"2019-12-26 10:12:53.000000000","updated_by":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"reviewer":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"state":"REVIEWER"},{"updated":"2019-12-26 10:19:28.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"5c6b79c85a210269a578dc5c7c9f656c186f7d6f","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2019-05-21 16:49:09.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"460c5a055d1f45c753424cfbfbb6a1d63e059e26","author":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"date":"2019-05-21 16:52:35.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"3c2a94d1333b571e16701e30dee8090b90d2ebf1","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-21 17:03:46.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-zuul-jobs-linters http://logs.openstack.org/54/660454/1/check/openstack-zuul-jobs-linters/9187fd2/ : SUCCESS in 5m 52s\n- project-config-gerrit http://logs.openstack.org/54/660454/1/check/project-config-gerrit/842a39f/ : SUCCESS in 3m 30s","accounts_in_message":[],"_revision_number":1},{"id":"af9c32ea7b76b5c8d33c10d6a509b6ab663c3337","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2019-05-21 19:48:33.000000000","message":"Patch Set 1: Workflow-1\n\nremvoe the legacy job from openstack-zuul-jobs as well - I\u0027ll only +2 once that is done. Therefore marking WIP.","accounts_in_message":[],"_revision_number":1},{"id":"e88f636d7335e4f90853d8837ceeb9fc737fd86e","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2019-05-21 20:08:58.000000000","message":"Patch Set 1:\n\nSubmitted https://review.opendev.org/660505","accounts_in_message":[],"_revision_number":1},{"id":"140e55a5e0dd5fd3cb62f039be876e856d9c0141","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2019-12-24 08:54:39.000000000","message":"Patch Set 1:\n\nDependency merged.\n\nhttps://review.opendev.org/#/q/project:x/devstack-plugin-sheepdog+is:open shows still two changes","accounts_in_message":[],"_revision_number":1},{"id":"e242419a1a72ed7dbc1a819f5e9eb5f82a7b80aa","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2019-12-26 10:12:53.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\ndependencies merged, everything abandoned.","accounts_in_message":[],"_revision_number":1},{"id":"5847447f89be10cd0c3e2fa219e0d9ac208e4d1e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-26 10:13:14.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"290bcf3b59c4fd49968b44b93dd3a60106d63e6c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-26 10:19:28.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/85d7a8100783459ca7b46fab512c6718 : SUCCESS in 6m 00s\n- project-config-gerrit https://zuul.opendev.org/t/openstack/build/459e78d4ffd24c2596c11313114a95ab : SUCCESS in 3m 22s","accounts_in_message":[],"_revision_number":1},{"id":"a20518a961d45a1b8561131b69bcb4575da546ee","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-26 10:19:29.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"ca79d057ff15296bd639c1647142aff5287dbea4","revisions":{"ca79d057ff15296bd639c1647142aff5287dbea4":{"kind":"REWORK","_number":1,"created":"2019-05-21 16:49:09.000000000","uploader":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"ref":"refs/changes/54/660454/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/54/660454/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/54/660454/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/54/660454/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/54/660454/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/54/660454/1"}}},"commit":{"parents":[{"commit":"787713c1e8efb5270c233161489b1e91ee124293","subject":"Drop jobs for devstack-plugins-sheepdog","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/787713c1e8efb5270c233161489b1e91ee124293"}]}],"author":{"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","date":"2019-05-21 16:48:06.000000000","tz":-300},"committer":{"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","date":"2019-05-21 16:48:06.000000000","tz":-300},"subject":"Remove devstack-plugin-sheepdog","message":"Remove devstack-plugin-sheepdog\n\nThis is step 4 in retiring the devstack-plugin-sheepdog project.\n\nDepends-on: https://review.opendev.org/#/c/660453/\nChange-Id: I16edd504036a1bc0f9510db50b94a4eaa0dc7518\nSigned-off-by: Sean McGinnis \u003csean.mcginnis@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/ca79d057ff15296bd639c1647142aff5287dbea4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/ca79d057ff15296bd639c1647142aff5287dbea4"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
