)]}'
{"id":"openstack%2Fproject-config~661187","triplet_id":"openstack%2Fproject-config~master~I631e3e3592a9be5c7da642cf34eec6d9758bf161","project":"openstack/project-config","branch":"master","topic":"letsencrypt","hashtags":[],"change_id":"I631e3e3592a9be5c7da642cf34eec6d9758bf161","subject":"Switch rax.dfw mirror to opendev.org","status":"MERGED","created":"2019-05-24 06:07:33.000000000","updated":"2019-05-28 20:05:42.000000000","submitted":"2019-05-28 20:05:41.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"661187-1559073942182-7cc59f37","meta_rev_id":"46723a321d0565ddc3043d2b41bbd65c86c6e6b2","_number":661187,"virtual_id_number":661187,"owner":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":1,"name":"James E. Blair","email":"jim@acmegating.com","username":"corvus"},{"value":0,"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":2,"date":"2019-05-28 20:05:41.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":1,"name":"James E. Blair","email":"jim@acmegating.com","username":"corvus"},"all":[{"value":2,"date":"2019-05-28 19:53:34.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":1,"name":"James E. Blair","email":"jim@acmegating.com","username":"corvus"},{"value":2,"date":"2019-05-24 08:19:53.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":1,"name":"James E. Blair","email":"jim@acmegating.com","username":"corvus"},"all":[{"value":1,"date":"2019-05-28 19:53:34.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":1,"name":"James E. Blair","email":"jim@acmegating.com","username":"corvus"},{"value":0,"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1,"name":"James E. Blair","email":"jim@acmegating.com","username":"corvus"},{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-05-24 08:19:53.000000000","updated_by":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"reviewer":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"state":"REVIEWER"},{"updated":"2019-05-28 19:53:34.000000000","updated_by":{"_account_id":1,"name":"James E. Blair","email":"jim@acmegating.com","username":"corvus"},"reviewer":{"_account_id":1,"name":"James E. Blair","email":"jim@acmegating.com","username":"corvus"},"state":"REVIEWER"},{"updated":"2019-05-28 20:05:41.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"1267293188b2b76d6d2e5cadfea4f941294e6fab","author":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"date":"2019-05-24 06:07:33.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"c5f9e5739847a9cbee48640d41ecc0a7f2d10c5f","author":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"date":"2019-05-24 06:12:53.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"afb3fa53c7e48752ecc11d2ca6ede0367889f611","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2019-05-24 06:24:39.000000000","message":"Patch Set 1: Code-Review+2\n\nI\u0027m fine with this trial...","accounts_in_message":[],"_revision_number":1},{"id":"0b69fb6778501ec765ee1f12f845543dfa89694b","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-24 06:25:33.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-zuul-jobs-linters http://logs.openstack.org/87/661187/1/check/openstack-zuul-jobs-linters/bea98ea/ : SUCCESS in 8m 05s","accounts_in_message":[],"_revision_number":1},{"id":"b02c7c8ca68dec468e343034f750a5af94644354","author":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"date":"2019-05-24 07:20:45.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"241b1cd223cd554b14af16067432743817b5c47f","author":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"date":"2019-05-24 07:22:10.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"68277baf9d03aecffb234917987ad98f27d4724c","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-24 07:30:07.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-zuul-jobs-linters http://logs.openstack.org/87/661187/2/check/openstack-zuul-jobs-linters/3989b39/ : SUCCESS in 6m 13s","accounts_in_message":[],"_revision_number":2},{"id":"0bd439063650be48b5e9ba3a8c93bb15fb13ea0b","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2019-05-24 08:19:53.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"19caf4e27744e00223c2940dc33c4fedfeb96879","author":{"_account_id":1,"name":"James E. Blair","email":"jim@acmegating.com","username":"corvus"},"date":"2019-05-28 19:53:34.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"ace291daf2c886b99299aa4561aa15dc303556ad","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-28 19:53:47.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"60b63c81f96921abc4c76d826bc8891e94aa4b55","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-28 20:05:41.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-zuul-jobs-linters http://logs.openstack.org/87/661187/2/gate/openstack-zuul-jobs-linters/af1c02a/ : SUCCESS in 6m 23s","accounts_in_message":[],"_revision_number":2},{"id":"b197ed0b66a33d320800728a030cd8d9cd879e92","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-28 20:05:42.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"05c0d363df5c1f32338d2da76bb0ba0734401eb5","revisions":{"9ab64488d5b208d1424bc7349d8f93e235ad6ecd":{"kind":"REWORK","_number":1,"created":"2019-05-24 06:07:33.000000000","uploader":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"ref":"refs/changes/87/661187/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/87/661187/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/87/661187/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/87/661187/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/87/661187/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/87/661187/1"}}},"commit":{"parents":[{"commit":"9e997e171963a1ec9fd0ce8b360c9238c64926d3","subject":"Merge \"Switch bindep fallback testing to openSUSE Leap 15.x\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/9e997e171963a1ec9fd0ce8b360c9238c64926d3"}]}],"author":{"name":"Ian Wienand","email":"iwienand@redhat.com","date":"2019-05-24 06:04:47.000000000","tz":600},"committer":{"name":"Ian Wienand","email":"iwienand@redhat.com","date":"2019-05-24 06:04:47.000000000","tz":600},"subject":"Switch rax.dfw mirror to opendev.org","message":"Switch rax.dfw mirror to opendev.org\n\nWe now have an opendev.org mirror in rax.dfw region (see\nI26b9ed1e21e2111f48bc7ecc384880c274eed213 and related changes).\n\nSwitch this in for initial testing.  As mentioned, this can grow as we\nbring new mirrors online, but will eventually be removed once all\nopendev.org mirrors are active.\n\nChange-Id: I631e3e3592a9be5c7da642cf34eec6d9758bf161\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/9ab64488d5b208d1424bc7349d8f93e235ad6ecd"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/9ab64488d5b208d1424bc7349d8f93e235ad6ecd"}]},"branch":"refs/heads/master"},"05c0d363df5c1f32338d2da76bb0ba0734401eb5":{"kind":"REWORK","_number":2,"created":"2019-05-24 07:22:10.000000000","uploader":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"ref":"refs/changes/87/661187/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/87/661187/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/87/661187/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/87/661187/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/87/661187/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/87/661187/2"}}},"commit":{"parents":[{"commit":"9e997e171963a1ec9fd0ce8b360c9238c64926d3","subject":"Merge \"Switch bindep fallback testing to openSUSE Leap 15.x\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/9e997e171963a1ec9fd0ce8b360c9238c64926d3"}]}],"author":{"name":"Ian Wienand","email":"iwienand@redhat.com","date":"2019-05-24 06:04:47.000000000","tz":600},"committer":{"name":"Ian Wienand","email":"iwienand@redhat.com","date":"2019-05-24 07:21:53.000000000","tz":600},"subject":"Switch rax.dfw mirror to opendev.org","message":"Switch rax.dfw mirror to opendev.org\n\nWe now have an opendev.org mirror in rax.dfw region (see\nI26b9ed1e21e2111f48bc7ecc384880c274eed213 and related changes).\n\nSwitch this in for initial testing.  As mentioned, this can grow as we\nbring new mirrors online, but will eventually be removed once all\nopendev.org mirrors are active.\n\nChange-Id: I631e3e3592a9be5c7da642cf34eec6d9758bf161\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/05c0d363df5c1f32338d2da76bb0ba0734401eb5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/05c0d363df5c1f32338d2da76bb0ba0734401eb5"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
