)]}'
{"id":"openstack%2Fproject-config~661332","triplet_id":"openstack%2Fproject-config~master~I2f253b03b3815dedd6cc65a3a7c9f098026cec1a","project":"openstack/project-config","branch":"master","topic":"retire_tripleo_ui","hashtags":[],"change_id":"I2f253b03b3815dedd6cc65a3a7c9f098026cec1a","subject":"Retiring TripleO-UI","status":"MERGED","created":"2019-05-24 15:22:16.000000000","updated":"2019-06-02 16:14:53.000000000","submitted":"2019-06-02 16:14:53.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":5,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"661332-1559492093874-fc91e94c","meta_rev_id":"4e55acf31674a680b5e0aabf5a5271db7fa1a587","_number":661332,"virtual_id_number":661332,"owner":{"_account_id":10112,"name":"Jason E. Rist","email":"jrist@redhat.com","username":"jrist"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":17888,"name":"Florian Fuchs","email":"f@florianfuchs.com","username":"flfuchs"},{"value":0,"_account_id":10112,"name":"Jason E. Rist","email":"jrist@redhat.com","username":"jrist"},{"value":0,"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"value":0,"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},{"value":2,"date":"2019-06-02 16:14:53.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":30413,"name":"chaiguanglong","email":"chaigl@awcloud.com","username":"chaiguanglong"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"all":[{"value":0,"_account_id":17888,"name":"Florian Fuchs","email":"f@florianfuchs.com","username":"flfuchs"},{"value":0,"_account_id":10112,"name":"Jason E. Rist","email":"jrist@redhat.com","username":"jrist"},{"value":0,"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"value":2,"date":"2019-06-01 07:14:10.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":1,"date":"2019-06-01 14:35:43.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":2,"date":"2019-06-02 16:02:33.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":30413,"name":"chaiguanglong","email":"chaigl@awcloud.com","username":"chaiguanglong"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"all":[{"value":0,"_account_id":17888,"name":"Florian Fuchs","email":"f@florianfuchs.com","username":"flfuchs"},{"value":0,"_account_id":10112,"name":"Jason E. Rist","email":"jrist@redhat.com","username":"jrist"},{"value":0,"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"value":0,"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":1,"date":"2019-06-02 16:02:33.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":30413,"name":"chaiguanglong","email":"chaigl@awcloud.com","username":"chaiguanglong"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"_account_id":10112,"name":"Jason E. Rist","email":"jrist@redhat.com","username":"jrist"},{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"_account_id":17888,"name":"Florian Fuchs","email":"f@florianfuchs.com","username":"flfuchs"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":30413,"name":"chaiguanglong","email":"chaigl@awcloud.com","username":"chaiguanglong"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-05-24 15:24:36.000000000","updated_by":{"_account_id":17888,"name":"Florian Fuchs","email":"f@florianfuchs.com","username":"flfuchs"},"reviewer":{"_account_id":17888,"name":"Florian Fuchs","email":"f@florianfuchs.com","username":"flfuchs"},"state":"REVIEWER"},{"updated":"2019-05-30 19:56:05.000000000","updated_by":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"reviewer":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"state":"REVIEWER"},{"updated":"2019-06-01 05:04:20.000000000","updated_by":{"_account_id":30413,"name":"chaiguanglong","email":"chaigl@awcloud.com","username":"chaiguanglong"},"reviewer":{"_account_id":30413,"name":"chaiguanglong","email":"chaigl@awcloud.com","username":"chaiguanglong"},"state":"REVIEWER"},{"updated":"2019-06-01 07:14:10.000000000","updated_by":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"reviewer":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"state":"REVIEWER"},{"updated":"2019-06-01 14:35:43.000000000","updated_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"reviewer":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"state":"REVIEWER"},{"updated":"2019-06-02 16:02:33.000000000","updated_by":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"reviewer":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"state":"REVIEWER"},{"updated":"2019-06-02 16:14:53.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"f24e700492f36c46babd72b9f33fe4a97c68fad5","author":{"_account_id":10112,"name":"Jason E. Rist","email":"jrist@redhat.com","username":"jrist"},"date":"2019-05-24 15:22:16.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"d4ba28426fe52ab857accd1ca29bdf9ab31c84fc","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-24 15:22:34.000000000","message":"Patch Set 1: Verified-1\n\nZuul encountered a syntax error while parsing its configuration in the\nrepo openstack/project-config on branch master.  The error was:\n\n  while scanning for the next token\nfound character \u0027\\t\u0027 that cannot start any token\n  in \"openstack/project-config/zuul.d/projects.yaml@master\", line 5063, column 1","accounts_in_message":[],"_revision_number":1},{"id":"12ca49906a36ea018d7659e04d26ba9c2fcdc7b0","author":{"_account_id":17888,"name":"Florian Fuchs","email":"f@florianfuchs.com","username":"flfuchs"},"date":"2019-05-24 15:24:36.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"a1bb6b935a87e2548bdb401b0507c90b9e0a3c0f","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2019-05-24 15:25:33.000000000","message":"Patch Set 1: Code-Review-1\n\n(1 comment)\n\nPlease be more verbose and link to any messages, see https://docs.openstack.org/infra/manual/drivers.html#retiring-a-project\n\nAre you retiring the whole repo - or keep stable branches for now?\n\nPut that answer also in the commit message.","accounts_in_message":[],"_revision_number":1},{"id":"71c87573b424b199ba8acc8a0a709edb10231d5b","author":{"_account_id":10112,"name":"Jason E. Rist","email":"jrist@redhat.com","username":"jrist"},"date":"2019-05-24 15:30:20.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"9f7eabfeb31c1b640f12620646b1f049d3b3e83d","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2019-05-24 15:32:09.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"ccaf818fe7110f0887406b85bfbe9c7eb28f8424","author":{"_account_id":10112,"name":"Jason E. Rist","email":"jrist@redhat.com","username":"jrist"},"date":"2019-05-24 17:01:16.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"7aa78c1084351432c84f6f9c8a65a189d39baeae","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-24 17:01:36.000000000","message":"Patch Set 2: Verified-1\n\n(1 comment)\n\nZuul encountered a syntax error while parsing its configuration in the\nrepo openstack/project-config on branch master.  The error was:\n\n  expected a dictionary for dictionary value @ data[\u0027jobs\u0027]\n\nThe error appears in the following project stanza:\n\n  project:\n      name: openstack/tripleo-ui\n      jobs:\n        - noop:\n            branches: master\n\n  in \"openstack/project-config/zuul.d/projects.yaml@master\", line 5059, column 3","accounts_in_message":[],"_revision_number":2},{"id":"3a45c2999772151e5d12e396564bfca6d5a7d31b","author":{"_account_id":10112,"name":"Jason E. Rist","email":"jrist@redhat.com","username":"jrist"},"date":"2019-05-24 21:35:32.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"822bbdefc571489c0b942bd513e9ff515f1b3125","author":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"date":"2019-05-30 19:56:05.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"cdfbc4efc188d0ac7fc65877f0c25936ae75d5ca","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-30 20:11:17.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-zuul-jobs-linters http://logs.openstack.org/32/661332/3/check/openstack-zuul-jobs-linters/3b2485e/ : SUCCESS in 5m 44s","accounts_in_message":[],"_revision_number":3},{"id":"1f0e6dde59d3318966a034862a0a8d2ad22a6e55","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2019-05-31 06:33:56.000000000","message":"Patch Set 3:\n\nDon\u0027t you need ansible-role-tripleo-ui for stable branches and can only retire this once you remove tripleo-ui stable branches as well?","accounts_in_message":[],"_revision_number":3},{"id":"c67c26fa76036db971e6c1d28bc79342a2446798","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2019-05-31 21:45:03.000000000","message":"Patch Set 3: Code-Review+1","accounts_in_message":[],"_revision_number":3},{"id":"27bea73f45987dba9f4d6539ae6d1ada2ffdb608","author":{"_account_id":30413,"name":"chaiguanglong","email":"chaigl@awcloud.com","username":"chaiguanglong"},"date":"2019-06-01 05:04:20.000000000","message":"Patch Set 3: Code-Review+1","accounts_in_message":[],"_revision_number":3},{"id":"fa988839e6830dd58ff49380e32f0470f1dcf79e","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2019-06-01 07:11:03.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"a544ccbb09213b367cb46ad4b11ae0abac989212","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2019-06-01 07:11:40.000000000","message":"Patch Set 3:\n\nAS discussed: ansible-role-tripleo-ui is unused.\n\nLet me add to commit message.","accounts_in_message":[],"_revision_number":3},{"id":"77067da8818ffaef182746e322028f7c8f65fa5f","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2019-06-01 07:12:54.000000000","message":"Patch Set 4: Commit message was updated.","accounts_in_message":[],"_revision_number":4},{"id":"fe87080f9cea1822f65a3398f52c33d54382ef61","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2019-06-01 07:14:10.000000000","message":"Patch Set 4: Code-Review+2\n\nJason, please provide the other steps for ansible-role-tripleo-ui as well, see https://docs.openstack.org/infra/manual/drivers.html#retiring-a-project for all details","accounts_in_message":[],"_revision_number":4},{"id":"377e09a6f9cd3522603f6d947e21dc098ff15dae","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-01 07:18:55.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-zuul-jobs-linters http://logs.openstack.org/32/661332/4/check/openstack-zuul-jobs-linters/ad3614a/ : SUCCESS in 4m 57s","accounts_in_message":[],"_revision_number":4},{"id":"d28dd83de203c1c7d57010d2d62f5fe010d242eb","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2019-06-01 14:35:43.000000000","message":"Patch Set 4: Code-Review+1","accounts_in_message":[],"_revision_number":4},{"id":"02980b61c1e598a16afd86484d7e36dbcc07b38a","author":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"date":"2019-06-02 16:02:33.000000000","message":"Patch Set 4: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":4},{"id":"f2e641c4f52416c0f832379d1102e9cf491da009","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-02 16:02:51.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"c396a0b38935e54a6e86a7b078c2a2625c2e2914","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-02 16:14:53.000000000","message":"Patch Set 4: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-zuul-jobs-linters http://logs.openstack.org/32/661332/4/gate/openstack-zuul-jobs-linters/d6d2ad5/ : SUCCESS in 6m 08s","accounts_in_message":[],"_revision_number":4},{"id":"62b1c621cd5d2996eac3faa643e7207abe758dee","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-02 16:14:53.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"914f71763d9e176c888f39fa9e6aeb18456630e2","revisions":{"f336ef32121be5830afcf94869697aa6aa66decf":{"kind":"REWORK","_number":1,"created":"2019-05-24 15:22:16.000000000","uploader":{"_account_id":10112,"name":"Jason E. Rist","email":"jrist@redhat.com","username":"jrist"},"ref":"refs/changes/32/661332/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/32/661332/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/32/661332/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/32/661332/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/32/661332/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/32/661332/1"}}},"commit":{"parents":[{"commit":"194a722058f70b164685c980d5ff2fa9bac87b26","subject":"Merge \"Switch api publish jobs to tox-docs\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/194a722058f70b164685c980d5ff2fa9bac87b26"}]}],"author":{"name":"Jason E. Rist","email":"jrist@redhat.com","date":"2019-05-24 15:21:51.000000000","tz":-360},"committer":{"name":"Jason E. Rist","email":"jrist@redhat.com","date":"2019-05-24 15:22:15.000000000","tz":-360},"subject":"Retiring TripleO-UI","message":"Retiring TripleO-UI\n\nChange-Id: I2f253b03b3815dedd6cc65a3a7c9f098026cec1a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/f336ef32121be5830afcf94869697aa6aa66decf"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/f336ef32121be5830afcf94869697aa6aa66decf"}]},"branch":"refs/heads/master"},"1535bf17580724b62a943f9c2dd24c4938278968":{"kind":"REWORK","_number":2,"created":"2019-05-24 17:01:16.000000000","uploader":{"_account_id":10112,"name":"Jason E. Rist","email":"jrist@redhat.com","username":"jrist"},"ref":"refs/changes/32/661332/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/32/661332/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/32/661332/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/32/661332/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/32/661332/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/32/661332/2"}}},"commit":{"parents":[{"commit":"194a722058f70b164685c980d5ff2fa9bac87b26","subject":"Merge \"Switch api publish jobs to tox-docs\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/194a722058f70b164685c980d5ff2fa9bac87b26"}]}],"author":{"name":"Jason E. Rist","email":"jrist@redhat.com","date":"2019-05-24 15:21:51.000000000","tz":-360},"committer":{"name":"Jason E. Rist","email":"jrist@redhat.com","date":"2019-05-24 17:00:50.000000000","tz":-360},"subject":"Retiring TripleO-UI","message":"Retiring TripleO-UI\n\nRemoving master from jobs but keeping stable branches on.\n\nChange-Id: I2f253b03b3815dedd6cc65a3a7c9f098026cec1a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/1535bf17580724b62a943f9c2dd24c4938278968"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/1535bf17580724b62a943f9c2dd24c4938278968"}]},"branch":"refs/heads/master"},"474d33dcc1d89b755c8856f627ea8572d7630dd2":{"kind":"REWORK","_number":3,"created":"2019-05-30 19:56:05.000000000","uploader":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"ref":"refs/changes/32/661332/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/32/661332/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/32/661332/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/32/661332/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/32/661332/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/32/661332/3"}}},"commit":{"parents":[{"commit":"194a722058f70b164685c980d5ff2fa9bac87b26","subject":"Merge \"Switch api publish jobs to tox-docs\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/194a722058f70b164685c980d5ff2fa9bac87b26"}]}],"author":{"name":"Jason E. Rist","email":"jrist@redhat.com","date":"2019-05-24 15:21:51.000000000","tz":-360},"committer":{"name":"Clark Boylan","email":"clark.boylan@gmail.com","date":"2019-05-30 19:55:56.000000000","tz":-420},"subject":"Retiring TripleO-UI","message":"Retiring TripleO-UI\n\nRemoving master from jobs but keeping stable branches on.\n\nChange-Id: I2f253b03b3815dedd6cc65a3a7c9f098026cec1a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/474d33dcc1d89b755c8856f627ea8572d7630dd2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/474d33dcc1d89b755c8856f627ea8572d7630dd2"}]},"branch":"refs/heads/master"},"914f71763d9e176c888f39fa9e6aeb18456630e2":{"kind":"NO_CODE_CHANGE","_number":4,"created":"2019-06-01 07:12:54.000000000","uploader":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"ref":"refs/changes/32/661332/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/32/661332/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/32/661332/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/32/661332/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/32/661332/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/32/661332/4"}}},"commit":{"parents":[{"commit":"194a722058f70b164685c980d5ff2fa9bac87b26","subject":"Merge \"Switch api publish jobs to tox-docs\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/194a722058f70b164685c980d5ff2fa9bac87b26"}]}],"author":{"name":"Jason E. Rist","email":"jrist@redhat.com","date":"2019-05-24 15:21:51.000000000","tz":-360},"committer":{"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","date":"2019-06-01 07:12:49.000000000","tz":0},"subject":"Retiring TripleO-UI","message":"Retiring TripleO-UI\n\nRemoving master from jobs but keeping stable branches on.\n\nRetire completely ansible-role-tripleo-ui, it\u0027s unused.\n\nChange-Id: I2f253b03b3815dedd6cc65a3a7c9f098026cec1a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/914f71763d9e176c888f39fa9e6aeb18456630e2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/914f71763d9e176c888f39fa9e6aeb18456630e2"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
