)]}'
{"id":"openstack%2Fproject-config~661455","triplet_id":"openstack%2Fproject-config~master~I644cd07ccda296a588fd5e897db0fcf71a3edaa1","project":"openstack/project-config","branch":"master","hashtags":[],"change_id":"I644cd07ccda296a588fd5e897db0fcf71a3edaa1","subject":"Update git load balancer for OpenDev","status":"MERGED","created":"2019-05-25 13:52:01.000000000","updated":"2019-05-25 14:16:08.000000000","submitted":"2019-05-25 14:16:08.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"661455-1558793768995-3080c1e4","meta_rev_id":"bcc8f66cdfde5549a9fc2d6d3a2446c34578d7f9","_number":661455,"virtual_id_number":661455,"owner":{"_account_id":1,"name":"James E. Blair","email":"jim@acmegating.com","username":"corvus"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":4162,"name":"Paul Belanger","email":"pabelanger@redhat.com","username":"pabelanger"},{"value":0,"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"value":2,"date":"2019-05-25 14:16:08.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":4162,"name":"Paul Belanger","email":"pabelanger@redhat.com","username":"pabelanger"},"all":[{"value":2,"date":"2019-05-25 13:55:51.000000000","_account_id":4162,"name":"Paul Belanger","email":"pabelanger@redhat.com","username":"pabelanger"},{"value":2,"date":"2019-05-25 13:58:54.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"all":[{"value":0,"_account_id":4162,"name":"Paul Belanger","email":"pabelanger@redhat.com","username":"pabelanger"},{"value":1,"date":"2019-05-25 13:58:54.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"_account_id":4162,"name":"Paul Belanger","email":"pabelanger@redhat.com","username":"pabelanger"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-05-25 13:55:51.000000000","updated_by":{"_account_id":4162,"name":"Paul Belanger","email":"pabelanger@redhat.com","username":"pabelanger"},"reviewer":{"_account_id":4162,"name":"Paul Belanger","email":"pabelanger@redhat.com","username":"pabelanger"},"state":"REVIEWER"},{"updated":"2019-05-25 13:58:54.000000000","updated_by":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"reviewer":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"state":"REVIEWER"},{"updated":"2019-05-25 14:16:08.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"f4e21fa866181ce7493dd4a047cf1b77f2236629","author":{"_account_id":1,"name":"James E. Blair","email":"jim@acmegating.com","username":"corvus"},"date":"2019-05-25 13:52:01.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"4b16a02620066eb80f3586a9ba67b19b1598491a","author":{"_account_id":4162,"name":"Paul Belanger","email":"pabelanger@redhat.com","username":"pabelanger"},"date":"2019-05-25 13:53:02.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"10556c08fa17ddc6010f20c974a24de89a707f49","author":{"_account_id":1,"name":"James E. Blair","email":"jim@acmegating.com","username":"corvus"},"date":"2019-05-25 13:55:05.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"8b339a54f1407d4de12af4014adff5add571034f","author":{"_account_id":4162,"name":"Paul Belanger","email":"pabelanger@redhat.com","username":"pabelanger"},"date":"2019-05-25 13:55:51.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"92cc1bc7a8e0f1fea787aea3ea7fd78cea69e921","author":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"date":"2019-05-25 13:58:54.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"185ae1a334f5f229532b62126a0b70131a1b11be","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-25 14:06:56.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-zuul-jobs-linters http://logs.openstack.org/55/661455/2/check/openstack-zuul-jobs-linters/4996544/ : SUCCESS in 5m 54s\n- project-config-grafyaml http://logs.openstack.org/55/661455/2/check/project-config-grafyaml/15d60bd/ : SUCCESS in 2m 44s","accounts_in_message":[],"_revision_number":2},{"id":"61c4edec3f02ccf567d70cc45d5ca4c0014c58d7","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-25 14:07:09.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"4ede0963e2a4dd9656a34806bf264e01ac431996","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-25 14:16:08.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-zuul-jobs-linters http://logs.openstack.org/55/661455/2/gate/openstack-zuul-jobs-linters/65af892/ : SUCCESS in 6m 04s\n- project-config-grafyaml http://logs.openstack.org/55/661455/2/gate/project-config-grafyaml/e396473/ : SUCCESS in 3m 13s","accounts_in_message":[],"_revision_number":2},{"id":"bcc8f66cdfde5549a9fc2d6d3a2446c34578d7f9","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-25 14:16:08.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"d60c3a1d49ed65b67a4a3bc97472e1097e644ece","revisions":{"553ad0d9f3e7512630a2e85fd72809e48c0e487f":{"kind":"REWORK","_number":1,"created":"2019-05-25 13:52:01.000000000","uploader":{"_account_id":1,"name":"James E. Blair","email":"jim@acmegating.com","username":"corvus"},"ref":"refs/changes/55/661455/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/55/661455/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/55/661455/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/55/661455/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/55/661455/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/55/661455/1"}}},"commit":{"parents":[{"commit":"82b3153ad236344ce71b4476843c894c55f3b699","subject":"Remove in-tree job from networking-bigswitch","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/82b3153ad236344ce71b4476843c894c55f3b699"}]}],"author":{"name":"James E. Blair","email":"jeblair@redhat.com","date":"2019-05-25 13:50:51.000000000","tz":-420},"committer":{"name":"James E. Blair","email":"jeblair@redhat.com","date":"2019-05-25 13:50:51.000000000","tz":-420},"subject":"Update git load balancer for OpenDev","message":"Update git load balancer for OpenDev\n\nThe git0X.openstack.org backends have been replaced by\ngitea0X.opendev.org.  Update the metric names to match, as well\nas the title of the dashboard.\n\nChange-Id: I644cd07ccda296a588fd5e897db0fcf71a3edaa1\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/553ad0d9f3e7512630a2e85fd72809e48c0e487f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/553ad0d9f3e7512630a2e85fd72809e48c0e487f"}]},"branch":"refs/heads/master"},"d60c3a1d49ed65b67a4a3bc97472e1097e644ece":{"kind":"REWORK","_number":2,"created":"2019-05-25 13:55:05.000000000","uploader":{"_account_id":1,"name":"James E. Blair","email":"jim@acmegating.com","username":"corvus"},"ref":"refs/changes/55/661455/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/55/661455/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/55/661455/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/55/661455/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/55/661455/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/55/661455/2"}}},"commit":{"parents":[{"commit":"82b3153ad236344ce71b4476843c894c55f3b699","subject":"Remove in-tree job from networking-bigswitch","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/82b3153ad236344ce71b4476843c894c55f3b699"}]}],"author":{"name":"James E. Blair","email":"jeblair@redhat.com","date":"2019-05-25 13:50:51.000000000","tz":-420},"committer":{"name":"James E. Blair","email":"jeblair@redhat.com","date":"2019-05-25 13:54:43.000000000","tz":-420},"subject":"Update git load balancer for OpenDev","message":"Update git load balancer for OpenDev\n\nThe git0X.openstack.org backends have been replaced by\ngitea0X.opendev.org.  Update the metric names to match, as well\nas the title of the dashboard.  Remove the git protocol section\nsince we don\u0027t support that anymore.\n\nChange-Id: I644cd07ccda296a588fd5e897db0fcf71a3edaa1\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/d60c3a1d49ed65b67a4a3bc97472e1097e644ece"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/d60c3a1d49ed65b67a4a3bc97472e1097e644ece"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
