)]}'
{"id":"openstack%2Fproject-config~662073","triplet_id":"openstack%2Fproject-config~master~Ie4ee3f9350d027f4cf53997aaa46a6099a512107","project":"openstack/project-config","branch":"master","topic":"castellan-ui-retirement","hashtags":[],"change_id":"Ie4ee3f9350d027f4cf53997aaa46a6099a512107","subject":"Retire castellan-ui","status":"MERGED","created":"2019-05-29 19:31:18.000000000","updated":"2019-05-29 21:05:08.000000000","submitted":"2019-05-29 21:05:08.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":3,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"662073-1559163908893-ac06e723","meta_rev_id":"ab0259e7ae483bb2548bdff7766a64aa7b2c793e","_number":662073,"virtual_id_number":662073,"owner":{"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":1,"name":"James E. Blair","email":"jim@acmegating.com","username":"corvus"},{"value":0,"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":0,"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},{"value":2,"date":"2019-05-29 21:05:08.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":1,"name":"James E. Blair","email":"jim@acmegating.com","username":"corvus"},"all":[{"value":2,"date":"2019-05-29 20:56:27.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":1,"name":"James E. Blair","email":"jim@acmegating.com","username":"corvus"},{"value":2,"date":"2019-05-29 19:53:55.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":0,"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":1,"name":"James E. Blair","email":"jim@acmegating.com","username":"corvus"},"all":[{"value":1,"date":"2019-05-29 20:56:27.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":1,"name":"James E. Blair","email":"jim@acmegating.com","username":"corvus"},{"value":0,"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":0,"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1,"name":"James E. Blair","email":"jim@acmegating.com","username":"corvus"},{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-05-29 19:53:55.000000000","updated_by":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"reviewer":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"state":"REVIEWER"},{"updated":"2019-05-29 20:56:27.000000000","updated_by":{"_account_id":1,"name":"James E. Blair","email":"jim@acmegating.com","username":"corvus"},"reviewer":{"_account_id":1,"name":"James E. Blair","email":"jim@acmegating.com","username":"corvus"},"state":"REVIEWER"},{"updated":"2019-05-29 21:05:08.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"7edff7ccb18531174d2f7e59e01bcf44a63aee02","author":{"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},"date":"2019-05-29 19:31:18.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"4f71ca16e8e985f9e9fe7663061eba4e9c1ebfd8","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-29 19:43:36.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-zuul-jobs-linters http://logs.openstack.org/73/662073/1/check/openstack-zuul-jobs-linters/77771a6/ : SUCCESS in 6m 19s\n- project-config-gerrit http://logs.openstack.org/73/662073/1/check/project-config-gerrit/be2821f/ : SUCCESS in 4m 19s","accounts_in_message":[],"_revision_number":1},{"id":"5a28412a43001ccc4f0c5481ab8e423c269fe49f","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2019-05-29 19:43:53.000000000","message":"Patch Set 1: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"abd4036b999d4513626a3c9e56f6663b9f853b87","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2019-05-29 19:44:34.000000000","message":"Patch Set 1:\n\ndo you have a goverance change up as well?","accounts_in_message":[],"_revision_number":1},{"id":"bd724013f5848e3d6353093a74bd12ffbee3323a","author":{"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},"date":"2019-05-29 19:47:24.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"c278c94393d495400b12188dfba1c704129f5f0c","author":{"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},"date":"2019-05-29 19:48:34.000000000","message":"Patch Set 1:\n\n(1 comment)\n\nThanks for the review.  The governance patch is here: https://review.opendev.org/#/c/662077/","accounts_in_message":[],"_revision_number":1},{"id":"e91a1f3dc82eeb0b0a2f0c0ec151597a4d81372c","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2019-05-29 19:48:57.000000000","message":"Patch Set 2: Code-Review-1\n\n(1 comment)\n\nthat should be it..","accounts_in_message":[],"_revision_number":2},{"id":"d62f70c9c811a585b8876908722f48429a2f7a11","author":{"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},"date":"2019-05-29 19:50:32.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"020e39de2607bcf089eb2dfe3c4e90195565dbdd","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2019-05-29 19:53:55.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"084721d26f70f041f8d2d61ab7502e4b1f8908d5","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-29 20:04:19.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-zuul-jobs-linters http://logs.openstack.org/73/662073/3/check/openstack-zuul-jobs-linters/0422a49/ : SUCCESS in 6m 07s\n- project-config-gerrit http://logs.openstack.org/73/662073/3/check/project-config-gerrit/a24d98f/ : SUCCESS in 3m 22s","accounts_in_message":[],"_revision_number":3},{"id":"bd2e58086530befc4e59a734d72bf7fccd3c1b0e","author":{"_account_id":1,"name":"James E. Blair","email":"jim@acmegating.com","username":"corvus"},"date":"2019-05-29 20:56:27.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"b1833e52745d007fb7dfc314c388d7b98ce233a5","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-29 20:57:04.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"95eb78b7dc28e2a368daf0057387a0f8f8a940c8","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-29 21:05:08.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-zuul-jobs-linters http://logs.openstack.org/73/662073/3/gate/openstack-zuul-jobs-linters/32cc6ca/ : SUCCESS in 5m 42s\n- project-config-gerrit http://logs.openstack.org/73/662073/3/gate/project-config-gerrit/f6e7256/ : SUCCESS in 3m 05s","accounts_in_message":[],"_revision_number":3},{"id":"403e07aadfd1c9679991fe3a864cd06916d33375","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-29 21:05:08.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"10f74af947e03cbb291aed7a952b2d666fedcc45","revisions":{"d75b844056a4e34b439cefd1d9c3ec9782e90ddd":{"kind":"REWORK","_number":1,"created":"2019-05-29 19:31:18.000000000","uploader":{"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},"ref":"refs/changes/73/662073/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/73/662073/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/73/662073/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/73/662073/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/73/662073/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/73/662073/1"}}},"commit":{"parents":[{"commit":"01b1cca0902a122173ca221f6e8ab5b27b5d4e64","subject":"Merge \"Retire x/pbrx and openstack-infra/opendev-website (Step 1)\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/01b1cca0902a122173ca221f6e8ab5b27b5d4e64"}]}],"author":{"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","date":"2019-05-29 19:29:09.000000000","tz":-300},"committer":{"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","date":"2019-05-29 19:29:09.000000000","tz":-300},"subject":"Retire castellan-ui","message":"Retire castellan-ui\n\ncastellan-ui is being retired. [1]  This patch removes the project\nfrom infrastructure systems (Step 4 in [2]).\n\n[1] http://lists.openstack.org/pipermail/openstack-discuss/2019-May/006509.html\n[2] https://docs.openstack.org/infra/manual/drivers.html#retiring-a-project\n\nChange-Id: Ie4ee3f9350d027f4cf53997aaa46a6099a512107\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/d75b844056a4e34b439cefd1d9c3ec9782e90ddd"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/d75b844056a4e34b439cefd1d9c3ec9782e90ddd"}]},"branch":"refs/heads/master"},"9c9bfe5f60cbfc18037fe8d2dd63a924d042906c":{"kind":"REWORK","_number":2,"created":"2019-05-29 19:47:24.000000000","uploader":{"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},"ref":"refs/changes/73/662073/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/73/662073/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/73/662073/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/73/662073/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/73/662073/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/73/662073/2"}}},"commit":{"parents":[{"commit":"01b1cca0902a122173ca221f6e8ab5b27b5d4e64","subject":"Merge \"Retire x/pbrx and openstack-infra/opendev-website (Step 1)\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/01b1cca0902a122173ca221f6e8ab5b27b5d4e64"}]}],"author":{"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","date":"2019-05-29 19:29:09.000000000","tz":-300},"committer":{"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","date":"2019-05-29 19:47:17.000000000","tz":-300},"subject":"Retire castellan-ui","message":"Retire castellan-ui\n\ncastellan-ui is being retired. [1]  This patch removes the project\nfrom infrastructure systems (Step 4 in [2]).\n\n[1] http://lists.openstack.org/pipermail/openstack-discuss/2019-May/006509.html\n[2] https://docs.openstack.org/infra/manual/drivers.html#retiring-a-project\n\nChange-Id: Ie4ee3f9350d027f4cf53997aaa46a6099a512107\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/9c9bfe5f60cbfc18037fe8d2dd63a924d042906c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/9c9bfe5f60cbfc18037fe8d2dd63a924d042906c"}]},"branch":"refs/heads/master"},"10f74af947e03cbb291aed7a952b2d666fedcc45":{"kind":"REWORK","_number":3,"created":"2019-05-29 19:50:32.000000000","uploader":{"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},"ref":"refs/changes/73/662073/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/73/662073/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/73/662073/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/73/662073/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/73/662073/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/73/662073/3"}}},"commit":{"parents":[{"commit":"01b1cca0902a122173ca221f6e8ab5b27b5d4e64","subject":"Merge \"Retire x/pbrx and openstack-infra/opendev-website (Step 1)\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/01b1cca0902a122173ca221f6e8ab5b27b5d4e64"}]}],"author":{"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","date":"2019-05-29 19:29:09.000000000","tz":-300},"committer":{"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","date":"2019-05-29 19:50:26.000000000","tz":-300},"subject":"Retire castellan-ui","message":"Retire castellan-ui\n\ncastellan-ui is being retired. [1]  This patch removes the project\nfrom infrastructure systems (Step 4 in [2]).\n\n[1] http://lists.openstack.org/pipermail/openstack-discuss/2019-May/006509.html\n[2] https://docs.openstack.org/infra/manual/drivers.html#retiring-a-project\n\nChange-Id: Ie4ee3f9350d027f4cf53997aaa46a6099a512107\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/10f74af947e03cbb291aed7a952b2d666fedcc45"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/10f74af947e03cbb291aed7a952b2d666fedcc45"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
