)]}'
{"id":"openstack%2Fproject-config~662223","triplet_id":"openstack%2Fproject-config~master~Ibabbe59b65eb52b32aa636487450b8ae1dd89269","project":"openstack/project-config","branch":"master","topic":"retired-acls","hashtags":[],"change_id":"Ibabbe59b65eb52b32aa636487450b8ae1dd89269","subject":"Use retired ACL everywhere","status":"MERGED","created":"2019-05-30 13:46:44.000000000","updated":"2019-05-31 23:43:54.000000000","submitted":"2019-05-31 23:43:54.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"662223-1559346234625-5004bf0a","meta_rev_id":"e462aa3a5b8cb2824d743acf807241600269ff14","_number":662223,"virtual_id_number":662223,"owner":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":1,"name":"James E. Blair","email":"jim@acmegating.com","username":"corvus"},{"value":0,"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"value":0,"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":2,"date":"2019-05-31 23:43:54.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":1,"name":"James E. Blair","email":"jim@acmegating.com","username":"corvus"},"all":[{"value":2,"date":"2019-05-31 23:21:04.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":1,"name":"James E. Blair","email":"jim@acmegating.com","username":"corvus"},{"value":2,"date":"2019-05-31 23:14:57.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"value":0,"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":1,"name":"James E. Blair","email":"jim@acmegating.com","username":"corvus"},"all":[{"value":1,"date":"2019-05-31 23:21:04.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":1,"name":"James E. Blair","email":"jim@acmegating.com","username":"corvus"},{"value":0,"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"value":0,"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1,"name":"James E. Blair","email":"jim@acmegating.com","username":"corvus"},{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-05-31 23:14:57.000000000","updated_by":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"reviewer":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"state":"REVIEWER"},{"updated":"2019-05-31 23:21:04.000000000","updated_by":{"_account_id":1,"name":"James E. Blair","email":"jim@acmegating.com","username":"corvus"},"reviewer":{"_account_id":1,"name":"James E. Blair","email":"jim@acmegating.com","username":"corvus"},"state":"REVIEWER"},{"updated":"2019-05-31 23:43:54.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"1eba0bc2a7079ae4dbd2ee1a39b561d78c9b7b14","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2019-05-30 13:46:44.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"155759bfe4c852c58c3cfa4dd63a8c03d4fcdcdc","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-30 13:59:01.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-zuul-jobs-linters http://logs.openstack.org/23/662223/1/check/openstack-zuul-jobs-linters/74a3019/ : SUCCESS in 6m 23s\n- project-config-gerrit http://logs.openstack.org/23/662223/1/check/project-config-gerrit/63d50e7/ : FAILURE in 2m 56s","accounts_in_message":[],"_revision_number":1},{"id":"28b956383d863381ffe286f6be29c6f68dacff90","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2019-05-30 14:09:28.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"5bfa929429deee275bdb19b2c329aede260dbdca","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-30 14:20:08.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-zuul-jobs-linters http://logs.openstack.org/23/662223/2/check/openstack-zuul-jobs-linters/7c04840/ : SUCCESS in 4m 56s\n- project-config-gerrit http://logs.openstack.org/23/662223/2/check/project-config-gerrit/f2cdc82/ : SUCCESS in 3m 22s","accounts_in_message":[],"_revision_number":2},{"id":"c18e846c29404c0b3eae87e3fe573e9c9b499722","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2019-05-31 15:46:56.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"946db62a6d38eb9a64de1eabb895c54ca22eef58","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-31 16:12:00.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-zuul-jobs-linters http://logs.openstack.org/23/662223/3/check/openstack-zuul-jobs-linters/8be871d/ : FAILURE in 4m 46s\n- project-config-gerrit http://logs.openstack.org/23/662223/3/check/project-config-gerrit/f5ec853/ : FAILURE in 2m 50s","accounts_in_message":[],"_revision_number":3},{"id":"db0b3d193a5becd3bcf988be222aabd903b63163","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2019-05-31 16:45:27.000000000","message":"Patch Set 3:\n\nrecheck","accounts_in_message":[],"_revision_number":3},{"id":"258aa7388efa4e1f20f7ee6c0ccd4e809e68ae52","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-31 16:55:08.000000000","message":"Patch Set 3:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-zuul-jobs-linters http://logs.openstack.org/23/662223/3/check/openstack-zuul-jobs-linters/3dc7fb1/ : SUCCESS in 5m 32s\n- project-config-gerrit http://logs.openstack.org/23/662223/3/check/project-config-gerrit/852553c/ : FAILURE in 2m 52s","accounts_in_message":[],"_revision_number":3},{"id":"44ee5a03a67cdc72c50db02995cc61a62f424bc3","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2019-05-31 18:55:21.000000000","message":"Patch Set 4: Patch Set 3 was rebased","accounts_in_message":[],"_revision_number":4},{"id":"d27ffb4a3b620f59a128750a48c43e06f82333d4","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-31 19:06:27.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-zuul-jobs-linters http://logs.openstack.org/23/662223/4/check/openstack-zuul-jobs-linters/65a4f7e/ : SUCCESS in 5m 58s\n- project-config-gerrit http://logs.openstack.org/23/662223/4/check/project-config-gerrit/e55ef11/ : SUCCESS in 2m 50s","accounts_in_message":[],"_revision_number":4},{"id":"70b135d46e25fdf8cfb2b0928e5d00199a03fd3c","author":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"date":"2019-05-31 23:14:57.000000000","message":"Patch Set 4: Code-Review+2","accounts_in_message":[],"_revision_number":4},{"id":"783472e6872f497c047bb7a36349eac7ee7cf4c4","author":{"_account_id":1,"name":"James E. Blair","email":"jim@acmegating.com","username":"corvus"},"date":"2019-05-31 23:21:04.000000000","message":"Patch Set 4: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":4},{"id":"b2a2565db63a8186ae5ee269cb3814feed6805af","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-31 23:21:18.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"b2c190b3a03560a375424eeb65936b7f16adfc56","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-31 23:43:54.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":4},{"id":"2ea2dbd31d81dcedda5f1cea120a0f64211bcb1f","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-31 23:43:54.000000000","message":"Patch Set 4: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-zuul-jobs-linters http://logs.openstack.org/23/662223/4/gate/openstack-zuul-jobs-linters/3601b46/ : SUCCESS in 6m 08s\n- project-config-gerrit http://logs.openstack.org/23/662223/4/gate/project-config-gerrit/1937444/ : SUCCESS in 3m 31s","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"2a72b2c1f8c365c4af6b8dbdbe2a20cdbb962e33","revisions":{"c0e19304d354cb5b5c9b2ac17d088a9cb5ce9888":{"kind":"REWORK","_number":1,"created":"2019-05-30 13:46:44.000000000","uploader":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"ref":"refs/changes/23/662223/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/23/662223/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/23/662223/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/23/662223/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/23/662223/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/23/662223/1"}}},"commit":{"parents":[{"commit":"9023462978134993e2ac7c08cc78b4637f08d6a0","subject":"Fix mirror hostnames after opendev migration test","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/9023462978134993e2ac7c08cc78b4637f08d6a0"}]}],"author":{"name":"Andreas Jaeger","email":"aj@suse.com","date":"2019-05-30 13:45:28.000000000","tz":120},"committer":{"name":"Andreas Jaeger","email":"aj@suse.com","date":"2019-05-30 13:46:32.000000000","tz":120},"subject":"Use retired ACL everywhere","message":"Use retired ACL everywhere\n\nUse the shared retired ACL for all retired repos, remove thus obsolete\nACLs.\n\nChange-Id: Ibabbe59b65eb52b32aa636487450b8ae1dd89269\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/c0e19304d354cb5b5c9b2ac17d088a9cb5ce9888"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/c0e19304d354cb5b5c9b2ac17d088a9cb5ce9888"}]},"branch":"refs/heads/master"},"9f1ca214c275bea0c530e170dcee694ff1e077b0":{"kind":"REWORK","_number":2,"created":"2019-05-30 14:09:28.000000000","uploader":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"ref":"refs/changes/23/662223/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/23/662223/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/23/662223/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/23/662223/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/23/662223/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/23/662223/2"}}},"commit":{"parents":[{"commit":"9023462978134993e2ac7c08cc78b4637f08d6a0","subject":"Fix mirror hostnames after opendev migration test","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/9023462978134993e2ac7c08cc78b4637f08d6a0"}]}],"author":{"name":"Andreas Jaeger","email":"aj@suse.com","date":"2019-05-30 13:45:28.000000000","tz":120},"committer":{"name":"Andreas Jaeger","email":"aj@suse.com","date":"2019-05-30 14:09:17.000000000","tz":120},"subject":"Use retired ACL everywhere","message":"Use retired ACL everywhere\n\nUse the shared retired ACL for all retired repos, remove thus obsolete\nACLs.\n\nChange-Id: Ibabbe59b65eb52b32aa636487450b8ae1dd89269\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/9f1ca214c275bea0c530e170dcee694ff1e077b0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/9f1ca214c275bea0c530e170dcee694ff1e077b0"}]},"branch":"refs/heads/master"},"c84e790dd2c7bc4d395fb9d6f982fff506ff9c8a":{"kind":"REWORK","_number":3,"created":"2019-05-31 15:46:56.000000000","uploader":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"ref":"refs/changes/23/662223/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/23/662223/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/23/662223/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/23/662223/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/23/662223/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/23/662223/3"}}},"commit":{"parents":[{"commit":"cd7622224f0fe209a6bee1385144c022f2335fce","subject":"Merge \"Rename x/charm-vault to openstack/charm-vault\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/cd7622224f0fe209a6bee1385144c022f2335fce"}]}],"author":{"name":"Andreas Jaeger","email":"aj@suse.com","date":"2019-05-30 13:45:28.000000000","tz":120},"committer":{"name":"Andreas Jaeger","email":"aj@suse.com","date":"2019-05-31 15:46:31.000000000","tz":120},"subject":"Use retired ACL everywhere","message":"Use retired ACL everywhere\n\nUse the shared retired ACL for all retired repos, remove thus obsolete\nACLs.\n\nChange-Id: Ibabbe59b65eb52b32aa636487450b8ae1dd89269\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/c84e790dd2c7bc4d395fb9d6f982fff506ff9c8a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/c84e790dd2c7bc4d395fb9d6f982fff506ff9c8a"}]},"branch":"refs/heads/master"},"2a72b2c1f8c365c4af6b8dbdbe2a20cdbb962e33":{"kind":"TRIVIAL_REBASE","_number":4,"created":"2019-05-31 18:55:21.000000000","uploader":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"ref":"refs/changes/23/662223/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/23/662223/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/23/662223/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/23/662223/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/23/662223/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/23/662223/4"}}},"commit":{"parents":[{"commit":"ef188ee2f60c5ee72da0b756149d5b73a6865fcc","subject":"Fix gerrit/projects.yaml sort order","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/ef188ee2f60c5ee72da0b756149d5b73a6865fcc"}]}],"author":{"name":"Andreas Jaeger","email":"aj@suse.com","date":"2019-05-30 13:45:28.000000000","tz":120},"committer":{"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","date":"2019-05-31 18:55:21.000000000","tz":0},"subject":"Use retired ACL everywhere","message":"Use retired ACL everywhere\n\nUse the shared retired ACL for all retired repos, remove thus obsolete\nACLs.\n\nChange-Id: Ibabbe59b65eb52b32aa636487450b8ae1dd89269\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/2a72b2c1f8c365c4af6b8dbdbe2a20cdbb962e33"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/2a72b2c1f8c365c4af6b8dbdbe2a20cdbb962e33"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
