)]}'
{"id":"openstack%2Fproject-config~663173","triplet_id":"openstack%2Fproject-config~master~Ide7d1da67e59000a9f1dc89046ce01e1f22cb8f8","project":"openstack/project-config","branch":"master","hashtags":[],"change_id":"Ide7d1da67e59000a9f1dc89046ce01e1f22cb8f8","subject":"Update gerritbot channels for #ara","status":"MERGED","created":"2019-06-05 00:14:51.000000000","updated":"2019-06-05 08:23:36.000000000","submitted":"2019-06-05 08:23:36.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"663173-1559723016684-4bf1be21","meta_rev_id":"20409ff0c372740aa069eda3bb5b2462e88cff71","_number":663173,"virtual_id_number":663173,"owner":{"_account_id":9061,"name":"David Moreau Simard","email":"moi@dmsimard.com","username":"dmsimard"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":2,"date":"2019-06-05 08:23:36.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"all":[{"value":2,"date":"2019-06-05 04:34:23.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":2,"date":"2019-06-05 08:15:28.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"all":[{"value":0,"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":1,"date":"2019-06-05 08:15:28.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-06-05 04:34:23.000000000","updated_by":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"reviewer":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"state":"REVIEWER"},{"updated":"2019-06-05 08:15:28.000000000","updated_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"reviewer":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"state":"REVIEWER"},{"updated":"2019-06-05 08:23:36.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"2a84e3e318ce27adf648eec4ca5916d93acd991b","author":{"_account_id":9061,"name":"David Moreau Simard","email":"moi@dmsimard.com","username":"dmsimard"},"date":"2019-06-05 00:14:51.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"3fc23b8e769fe761f338e6c6593e0fabda6d9381","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-05 00:24:14.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-zuul-jobs-linters http://logs.openstack.org/73/663173/1/check/openstack-zuul-jobs-linters/475a6a4/ : SUCCESS in 5m 47s\n- project-config-irc-access http://logs.openstack.org/73/663173/1/check/project-config-irc-access/ea5547d/ : SUCCESS in 6m 08s","accounts_in_message":[],"_revision_number":1},{"id":"2133908b1fb2a3029bf35bf365665ae751b26bb4","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2019-06-05 04:34:23.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"bba7405e6768e3b836ce80316080bd10196c41ad","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2019-06-05 08:15:28.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"24e92ecd1de7f1414cc29ed0126bff0d555760d4","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-05 08:15:41.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"f868b2f999fbe0c157f0a0ac35b02fe7fe832182","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-05 08:23:36.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"20409ff0c372740aa069eda3bb5b2462e88cff71","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-05 08:23:36.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-zuul-jobs-linters http://logs.openstack.org/73/663173/1/gate/openstack-zuul-jobs-linters/def5526/ : SUCCESS in 6m 45s\n- project-config-irc-access http://logs.openstack.org/73/663173/1/gate/project-config-irc-access/545e972/ : SUCCESS in 6m 25s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"4047d7fd9cf2aabd73ccf9bfe25595ecf26cc3d6","revisions":{"4047d7fd9cf2aabd73ccf9bfe25595ecf26cc3d6":{"kind":"REWORK","_number":1,"created":"2019-06-05 00:14:51.000000000","uploader":{"_account_id":9061,"name":"David Moreau Simard","email":"moi@dmsimard.com","username":"dmsimard"},"ref":"refs/changes/73/663173/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/73/663173/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/73/663173/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/73/663173/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/73/663173/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/73/663173/1"}}},"commit":{"parents":[{"commit":"a12c6b531f58aaf9c838299cc0f2abc8c9ee9f40","subject":"Merge \"Build ubuntu bionic minimal images for control plane\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/a12c6b531f58aaf9c838299cc0f2abc8c9ee9f40"}]}],"author":{"name":"David Moreau Simard","email":"dmsimard@redhat.com","date":"2019-06-05 00:13:21.000000000","tz":-240},"committer":{"name":"David Moreau Simard","email":"dmsimard@redhat.com","date":"2019-06-05 00:14:50.000000000","tz":-240},"subject":"Update gerritbot channels for #ara","message":"Update gerritbot channels for #ara\n\nfeature/1.0 has moved to master and master has moved to stable/0.x.\n\nChange-Id: Ide7d1da67e59000a9f1dc89046ce01e1f22cb8f8\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/4047d7fd9cf2aabd73ccf9bfe25595ecf26cc3d6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/4047d7fd9cf2aabd73ccf9bfe25595ecf26cc3d6"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
