)]}'
{"id":"openstack%2Fproject-config~664617","triplet_id":"openstack%2Fproject-config~master~Id97aceb78019b7b71bc231778d7ea7e0f3964e0d","project":"openstack/project-config","branch":"master","topic":"dev/arnaud.morin/remove_glean_specific_config","hashtags":[],"change_id":"Id97aceb78019b7b71bc231778d7ea7e0f3964e0d","subject":"Remove glean config for OVH nodes","status":"MERGED","created":"2019-06-11 15:12:00.000000000","updated":"2019-06-11 15:56:53.000000000","submitted":"2019-06-11 15:56:53.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"664617-1560268613463-b5d80b2d","meta_rev_id":"500d92999cda974b128b7f46aff6e6b7e036c908","_number":664617,"virtual_id_number":664617,"owner":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"value":2,"date":"2019-06-11 15:56:53.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"date":"2019-06-11 15:12:53.000000000","_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},{"value":0,"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"all":[{"value":2,"date":"2019-06-11 15:47:40.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},{"value":2,"date":"2019-06-11 15:38:02.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"all":[{"value":1,"date":"2019-06-11 15:47:40.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},{"value":0,"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-06-11 15:38:02.000000000","updated_by":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"reviewer":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"state":"REVIEWER"},{"updated":"2019-06-11 15:47:40.000000000","updated_by":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"reviewer":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"state":"REVIEWER"},{"updated":"2019-06-11 15:56:53.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"963c6540b8bec54d972c430cb562a789a37826cf","author":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"date":"2019-06-11 15:12:00.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"72205fcddaa94b6ebf2a4f9a1288b43bc8a40c40","author":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"date":"2019-06-11 15:12:53.000000000","message":"Patch Set 1:\n\nFeel free to test it manually before merging it","accounts_in_message":[],"_revision_number":1},{"id":"83845d59f73b6bf060fffaf2b221598e9e5ba418","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-11 15:25:36.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-zuul-jobs-linters http://logs.openstack.org/17/664617/1/check/openstack-zuul-jobs-linters/6cf6471/ : SUCCESS in 7m 09s\n- project-config-nodepool http://logs.openstack.org/17/664617/1/check/project-config-nodepool/dcb5ee2/ : SUCCESS in 3m 51s","accounts_in_message":[],"_revision_number":1},{"id":"268726bec9355e28d0a508fc0f74d36447957ecb","author":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"date":"2019-06-11 15:38:02.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"e1648dad9e532002b9458e462818add2fceb3129","author":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"date":"2019-06-11 15:47:40.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"eaa4809cf192eb099fcbc569ae0f669ee08d2d67","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-11 15:47:53.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"b1fd39cca5f8433bc87347a1204bd0e1e113ac62","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-11 15:56:53.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-zuul-jobs-linters http://logs.openstack.org/17/664617/1/gate/openstack-zuul-jobs-linters/a80e1c1/ : SUCCESS in 7m 22s\n- project-config-nodepool http://logs.openstack.org/17/664617/1/gate/project-config-nodepool/dbb038d/ : SUCCESS in 5m 15s","accounts_in_message":[],"_revision_number":1},{"id":"0a5e2d65dd47290d3001e30bc59e3868ba5e363b","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-11 15:56:53.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"56a21b2622d2a36fbdc199d397dcb4512f2fe92e","revisions":{"56a21b2622d2a36fbdc199d397dcb4512f2fe92e":{"kind":"REWORK","_number":1,"created":"2019-06-11 15:12:00.000000000","uploader":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"ref":"refs/changes/17/664617/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/17/664617/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/17/664617/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/17/664617/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/17/664617/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/17/664617/1"}}},"commit":{"parents":[{"commit":"fcc30e132fdea03ebff2d819ff0fb77982707897","subject":"Add api-ref for Qinling","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/fcc30e132fdea03ebff2d819ff0fb77982707897"}]}],"author":{"name":"Arnaud Morin","email":"arnaud.morin@corp.ovh.com","date":"2019-06-11 14:55:28.000000000","tz":120},"committer":{"name":"Arnaud Morin","email":"arnaud.morin@corp.ovh.com","date":"2019-06-11 14:55:28.000000000","tz":120},"subject":"Remove glean config for OVH nodes","message":"Remove glean config for OVH nodes\n\nOVH updated its infra in order to have correct network_data.json given\nby metadata API and / or config drive.\n\nSo glean override for this is not needed anymore.\n\nChange-Id: Id97aceb78019b7b71bc231778d7ea7e0f3964e0d\nSigned-off-by: Arnaud Morin \u003carnaud.morin@corp.ovh.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/56a21b2622d2a36fbdc199d397dcb4512f2fe92e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/56a21b2622d2a36fbdc199d397dcb4512f2fe92e"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
