)]}'
{"id":"openstack%2Fproject-config~664641","triplet_id":"openstack%2Fproject-config~master~I1728359788f818f2b3c5b5942f6f63bbff4c02d9","project":"openstack/project-config","branch":"master","topic":"add-airship-election","hashtags":[],"change_id":"I1728359788f818f2b3c5b5942f6f63bbff4c02d9","subject":"Add airship/election","status":"MERGED","created":"2019-06-11 17:23:42.000000000","updated":"2019-06-18 04:44:32.000000000","submitted":"2019-06-18 04:44:32.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"664641-1560833072670-eb0806c3","meta_rev_id":"caf1feec3a4d63958f1758d80c0d73bb821a0c43","_number":664641,"virtual_id_number":664641,"owner":{"_account_id":22477,"name":"Matt McEuen","email":"matt.mceuen@att.com","username":"mattmceuen"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":2,"date":"2019-06-18 04:44:32.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":22477,"name":"Matt McEuen","email":"matt.mceuen@att.com","username":"mattmceuen"},{"value":0,"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},{"value":0,"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":22477,"name":"Matt McEuen","email":"matt.mceuen@att.com","username":"mattmceuen"},{"value":2,"date":"2019-06-18 04:34:25.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},{"value":2,"date":"2019-06-17 14:55:16.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":22477,"name":"Matt McEuen","email":"matt.mceuen@att.com","username":"mattmceuen"},{"value":1,"date":"2019-06-18 04:34:25.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},{"value":0,"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22477,"name":"Matt McEuen","email":"matt.mceuen@att.com","username":"mattmceuen"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-06-17 14:55:16.000000000","updated_by":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"reviewer":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"state":"REVIEWER"},{"updated":"2019-06-18 04:34:25.000000000","updated_by":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"reviewer":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"state":"REVIEWER"},{"updated":"2019-06-18 04:44:32.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"7a1f60bf6b036cd2311fa591922d6ca68edbef8c","author":{"_account_id":22477,"name":"Matt McEuen","email":"matt.mceuen@att.com","username":"mattmceuen"},"date":"2019-06-11 17:23:42.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"15397a31ea53581e0c90dbb745da63c20dac6381","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-11 17:39:18.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-zuul-jobs-linters http://logs.openstack.org/41/664641/1/check/openstack-zuul-jobs-linters/ef5207c/ : SUCCESS in 6m 36s\n- project-config-gerrit http://logs.openstack.org/41/664641/1/check/project-config-gerrit/912e43c/ : FAILURE in 3m 29s","accounts_in_message":[],"_revision_number":1},{"id":"06024fca2d79453da72be6768243024858c467ea","author":{"_account_id":22477,"name":"Matt McEuen","email":"matt.mceuen@att.com","username":"mattmceuen"},"date":"2019-06-11 18:27:56.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"f5697c90741dcd5554d57c062e787abd3ee085e0","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-11 18:39:16.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-zuul-jobs-linters http://logs.openstack.org/41/664641/1/check/openstack-zuul-jobs-linters/b9805b2/ : SUCCESS in 6m 12s\n- project-config-gerrit http://logs.openstack.org/41/664641/1/check/project-config-gerrit/604670a/ : FAILURE in 3m 29s","accounts_in_message":[],"_revision_number":1},{"id":"fb025d931305ce6fe41d95b2daabeb70d6c1c1ef","author":{"_account_id":22477,"name":"Matt McEuen","email":"matt.mceuen@att.com","username":"mattmceuen"},"date":"2019-06-11 19:01:16.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"643e64e8eb0a641a09baf56c717401297b1dfb8f","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-11 19:30:41.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-zuul-jobs-linters http://logs.openstack.org/41/664641/2/check/openstack-zuul-jobs-linters/2f57e2b/ : SUCCESS in 6m 22s\n- project-config-gerrit http://logs.openstack.org/41/664641/2/check/project-config-gerrit/68f3f14/ : SUCCESS in 4m 06s","accounts_in_message":[],"_revision_number":2},{"id":"28d73c8fee63ef629f1f094842bc20e20ca58e8b","author":{"_account_id":22477,"name":"Matt McEuen","email":"matt.mceuen@att.com","username":"mattmceuen"},"date":"2019-06-13 16:51:03.000000000","message":"Patch Set 3: Patch Set 2 was rebased","accounts_in_message":[],"_revision_number":3},{"id":"1c86eeae1741447af3f7b37ad5516b6f0762152b","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-13 17:19:06.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-zuul-jobs-linters http://logs.openstack.org/41/664641/3/check/openstack-zuul-jobs-linters/0eebec7/ : SUCCESS in 6m 03s\n- project-config-gerrit http://logs.openstack.org/41/664641/3/check/project-config-gerrit/347d737/ : SUCCESS in 4m 38s","accounts_in_message":[],"_revision_number":3},{"id":"b8086da795e8d6da935e329480e5a00ccbf34c7d","author":{"_account_id":22477,"name":"Matt McEuen","email":"matt.mceuen@att.com","username":"mattmceuen"},"date":"2019-06-14 14:26:13.000000000","message":"Topic changed from add-airship-election to new-project","accounts_in_message":[],"_revision_number":3},{"id":"7b01688f97e9d1b888f6395900a320339c3cc6bc","author":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"date":"2019-06-17 14:20:41.000000000","message":"Patch Set 3:\n\nAt a minimum you\u0027re probably also going to want to add it to the zuul/main.yaml untrusted-projects list for the openstack tenant, per https://docs.openstack.org/infra/manual/creators.html#add-project-to-zuul (otherwise you\u0027ll be unable to merge new changes).","accounts_in_message":[],"_revision_number":3},{"id":"59233eb4528dbf6f95e0169d28f9be0f7f20d904","author":{"_account_id":22477,"name":"Matt McEuen","email":"matt.mceuen@att.com","username":"mattmceuen"},"date":"2019-06-17 14:41:52.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"65d86a10b5d6aef4b72b0abca255f0c49441e467","author":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"date":"2019-06-17 14:55:16.000000000","message":"Patch Set 4: Code-Review+2","accounts_in_message":[],"_revision_number":4},{"id":"2f104967e1999805e9648fb8fee512a17dc3de82","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-17 15:02:50.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-zuul-jobs-linters http://logs.openstack.org/41/664641/4/check/openstack-zuul-jobs-linters/0edd68e/ : SUCCESS in 7m 04s\n- project-config-gerrit http://logs.openstack.org/41/664641/4/check/project-config-gerrit/4d43424/ : SUCCESS in 4m 20s","accounts_in_message":[],"_revision_number":4},{"id":"93d515a34a4ff3477477a362caf501bce0a0a4e9","author":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"date":"2019-06-18 04:34:25.000000000","message":"Patch Set 4: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":4},{"id":"a7f65229dd7f4fce07c43fd056ab8ee0ac12a35b","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-18 04:34:39.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"96dd958cd697f043d3669e57a9ba61c15dc0ed1f","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-18 04:44:32.000000000","message":"Patch Set 4: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-zuul-jobs-linters http://logs.openstack.org/41/664641/4/gate/openstack-zuul-jobs-linters/c5e47cd/ : SUCCESS in 7m 03s\n- project-config-gerrit http://logs.openstack.org/41/664641/4/gate/project-config-gerrit/f7039bf/ : SUCCESS in 3m 14s","accounts_in_message":[],"_revision_number":4},{"id":"c326614fd51376935ccfa35e164c7e8e82465f60","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-18 04:44:32.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"f82147cde8ecff6f9fa0a527ed63c016dbb47c49","revisions":{"fa6721e0dd4308f09bbb8e326b6f31f16e5d341f":{"kind":"REWORK","_number":1,"created":"2019-06-11 17:23:42.000000000","uploader":{"_account_id":22477,"name":"Matt McEuen","email":"matt.mceuen@att.com","username":"mattmceuen"},"ref":"refs/changes/41/664641/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/41/664641/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/41/664641/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/41/664641/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/41/664641/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/41/664641/1"}}},"commit":{"parents":[{"commit":"f19880c9fcb567c46206a282bf524f64106a8c88","subject":"Increase the size of ext4 journal on control plane images","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/f19880c9fcb567c46206a282bf524f64106a8c88"}]}],"author":{"name":"Matt McEuen","email":"matt.mceuen@att.com","date":"2019-06-10 18:19:53.000000000","tz":-300},"committer":{"name":"Matt McEuen","email":"matt.mceuen@att.com","date":"2019-06-10 18:24:11.000000000","tz":-300},"subject":"Add airship/election","message":"Add airship/election\n\nThis adds the Airship Election documentation project, which\nis modeled after the StarlingX and OpenStack Election projects.\n\nChange-Id: I1728359788f818f2b3c5b5942f6f63bbff4c02d9\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/fa6721e0dd4308f09bbb8e326b6f31f16e5d341f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/fa6721e0dd4308f09bbb8e326b6f31f16e5d341f"}]},"branch":"refs/heads/master"},"b4962d22615739be03179c244fe6158796001fc2":{"kind":"REWORK","_number":2,"created":"2019-06-11 19:01:16.000000000","uploader":{"_account_id":22477,"name":"Matt McEuen","email":"matt.mceuen@att.com","username":"mattmceuen"},"ref":"refs/changes/41/664641/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/41/664641/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/41/664641/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/41/664641/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/41/664641/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/41/664641/2"}}},"commit":{"parents":[{"commit":"f19880c9fcb567c46206a282bf524f64106a8c88","subject":"Increase the size of ext4 journal on control plane images","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/f19880c9fcb567c46206a282bf524f64106a8c88"}]}],"author":{"name":"Matt McEuen","email":"matt.mceuen@att.com","date":"2019-06-10 18:19:53.000000000","tz":-300},"committer":{"name":"Matt McEuen","email":"matt.mceuen@att.com","date":"2019-06-11 19:01:10.000000000","tz":-300},"subject":"Add airship/election","message":"Add airship/election\n\nThis adds the Airship Election documentation project, which\nis modeled after the StarlingX and OpenStack Election projects.\n\nChange-Id: I1728359788f818f2b3c5b5942f6f63bbff4c02d9\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/b4962d22615739be03179c244fe6158796001fc2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/b4962d22615739be03179c244fe6158796001fc2"}]},"branch":"refs/heads/master"},"bbec825895f86a99174b1174c125d6a53ee90084":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2019-06-13 16:51:03.000000000","uploader":{"_account_id":22477,"name":"Matt McEuen","email":"matt.mceuen@att.com","username":"mattmceuen"},"ref":"refs/changes/41/664641/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/41/664641/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/41/664641/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/41/664641/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/41/664641/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/41/664641/3"}}},"commit":{"parents":[{"commit":"4023e6cfb17b4c4b10976f29425c7983571ea3d3","subject":"Merge \"Retire independent ansible role\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/4023e6cfb17b4c4b10976f29425c7983571ea3d3"}]}],"author":{"name":"Matt McEuen","email":"matt.mceuen@att.com","date":"2019-06-10 18:19:53.000000000","tz":-300},"committer":{"name":"Matt McEuen","email":"matt.mceuen@att.com","date":"2019-06-13 16:51:03.000000000","tz":0},"subject":"Add airship/election","message":"Add airship/election\n\nThis adds the Airship Election documentation project, which\nis modeled after the StarlingX and OpenStack Election projects.\n\nChange-Id: I1728359788f818f2b3c5b5942f6f63bbff4c02d9\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/bbec825895f86a99174b1174c125d6a53ee90084"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/bbec825895f86a99174b1174c125d6a53ee90084"}]},"branch":"refs/heads/master"},"f82147cde8ecff6f9fa0a527ed63c016dbb47c49":{"kind":"REWORK","_number":4,"created":"2019-06-17 14:41:52.000000000","uploader":{"_account_id":22477,"name":"Matt McEuen","email":"matt.mceuen@att.com","username":"mattmceuen"},"ref":"refs/changes/41/664641/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/41/664641/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/41/664641/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/41/664641/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/41/664641/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/41/664641/4"}}},"commit":{"parents":[{"commit":"f19880c9fcb567c46206a282bf524f64106a8c88","subject":"Increase the size of ext4 journal on control plane images","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/f19880c9fcb567c46206a282bf524f64106a8c88"}]}],"author":{"name":"Matt McEuen","email":"matt.mceuen@att.com","date":"2019-06-10 18:19:53.000000000","tz":-300},"committer":{"name":"Matt McEuen","email":"matt.mceuen@att.com","date":"2019-06-17 14:40:53.000000000","tz":-300},"subject":"Add airship/election","message":"Add airship/election\n\nThis adds the Airship Election documentation project, which\nis modeled after the StarlingX and OpenStack Election projects.\n\nChange-Id: I1728359788f818f2b3c5b5942f6f63bbff4c02d9\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/f82147cde8ecff6f9fa0a527ed63c016dbb47c49"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/f82147cde8ecff6f9fa0a527ed63c016dbb47c49"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
