)]}'
{"id":"openstack%2Fproject-config~664932","triplet_id":"openstack%2Fproject-config~master~Ia22150be91e7f45f59be4bab381727faa3d637df","project":"openstack/project-config","branch":"master","topic":"retire-role","hashtags":[],"change_id":"Ia22150be91e7f45f59be4bab381727faa3d637df","subject":"Retire independent ansible role","status":"MERGED","created":"2019-06-12 13:54:14.000000000","updated":"2019-06-13 04:49:51.000000000","submitted":"2019-06-13 04:49:51.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"664932-1560401391750-8e310be5","meta_rev_id":"70b22e72cca61248ba1500b6c0e470d6b89fec65","_number":664932,"virtual_id_number":664932,"owner":{"_account_id":7353,"name":"Kevin Carter","email":"kevin@cloudnull.com","username":"cloudnull"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":7353,"name":"Kevin Carter","email":"kevin@cloudnull.com","username":"cloudnull"},{"value":2,"date":"2019-06-13 04:49:51.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},{"value":0,"_account_id":30127,"name":"John Hou","email":"houj@awcloud.com","username":"houj"},{"value":0,"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"recommended":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"all":[{"value":1,"date":"2019-06-12 15:16:37.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":7353,"name":"Kevin Carter","email":"kevin@cloudnull.com","username":"cloudnull"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2019-06-13 02:20:50.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},{"value":1,"date":"2019-06-13 02:23:24.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":30127,"name":"John Hou","email":"houj@awcloud.com","username":"houj"},{"value":2,"date":"2019-06-13 04:43:19.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"all":[{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":7353,"name":"Kevin Carter","email":"kevin@cloudnull.com","username":"cloudnull"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},{"value":0,"_account_id":30127,"name":"John Hou","email":"houj@awcloud.com","username":"houj"},{"value":1,"date":"2019-06-13 04:43:19.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},{"_account_id":7353,"name":"Kevin Carter","email":"kevin@cloudnull.com","username":"cloudnull"},{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":30127,"name":"John Hou","email":"houj@awcloud.com","username":"houj"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-06-12 15:16:37.000000000","updated_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"reviewer":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"state":"REVIEWER"},{"updated":"2019-06-13 02:20:50.000000000","updated_by":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"reviewer":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"state":"REVIEWER"},{"updated":"2019-06-13 02:23:24.000000000","updated_by":{"_account_id":30127,"name":"John Hou","email":"houj@awcloud.com","username":"houj"},"reviewer":{"_account_id":30127,"name":"John Hou","email":"houj@awcloud.com","username":"houj"},"state":"REVIEWER"},{"updated":"2019-06-13 04:43:19.000000000","updated_by":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"reviewer":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"state":"REVIEWER"},{"updated":"2019-06-13 04:49:51.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"88f4c191dd2777e7f975b4a7ec517eb1025644ef","author":{"_account_id":7353,"name":"Kevin Carter","email":"kevin@cloudnull.com","username":"cloudnull"},"date":"2019-06-12 13:54:14.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"6f2d8893234933617df107eaba0b0fafd8606326","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2019-06-12 14:03:25.000000000","message":"Patch Set 1: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"5d089fb01429310e3ee42f7f5a1341419489201f","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-12 14:04:51.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-zuul-jobs-linters http://logs.openstack.org/32/664932/1/check/openstack-zuul-jobs-linters/94fb60e/ : SUCCESS in 6m 58s","accounts_in_message":[],"_revision_number":1},{"id":"7579add67815553aa0f1ad06fb9586b1a691a119","author":{"_account_id":7353,"name":"Kevin Carter","email":"kevin@cloudnull.com","username":"cloudnull"},"date":"2019-06-12 14:31:52.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"bd8284f1977fb991f9cd552853e4e47037e0940a","author":{"_account_id":7353,"name":"Kevin Carter","email":"kevin@cloudnull.com","username":"cloudnull"},"date":"2019-06-12 15:11:07.000000000","message":"Patch Set 2: Published edit on patch set 1.","accounts_in_message":[],"_revision_number":2},{"id":"aad3f736c6a55e113dbd4d9f322ebbb70bef9be0","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2019-06-12 15:16:37.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"15359bb1ef5ddc69783c61471ba5abd621ed4257","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-12 15:23:04.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-zuul-jobs-linters http://logs.openstack.org/32/664932/2/check/openstack-zuul-jobs-linters/e083f56/ : SUCCESS in 5m 48s","accounts_in_message":[],"_revision_number":2},{"id":"9965ae78d0f29228ae56b6b82b6f07ac16d47a49","author":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"date":"2019-06-13 02:20:50.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"1f68147b63ea5dc7224d08ff2d18d450a164c18c","author":{"_account_id":30127,"name":"John Hou","email":"houj@awcloud.com","username":"houj"},"date":"2019-06-13 02:23:24.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"ac157c56b81003c51301ad9cad93ecb7b0f03452","author":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"date":"2019-06-13 04:43:19.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"faef9e93d00e5d8dcbd1acdaf3dbdd89ed12df70","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-13 04:43:37.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"67c1b1fbb1d687b04004ea91dbba6ba85344d32b","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-13 04:49:51.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"cda7c4baf0d036b754d236228c3eecf76f66edfb","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-13 04:49:51.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-zuul-jobs-linters http://logs.openstack.org/32/664932/2/gate/openstack-zuul-jobs-linters/6905ae6/ : SUCCESS in 5m 29s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"9716bcfd58f6ae875ed0b532bd22258a9fb1f4c3","revisions":{"d038fec288638df9de5bd40d65bebe6aedd3aa25":{"kind":"REWORK","_number":1,"created":"2019-06-12 13:54:14.000000000","uploader":{"_account_id":7353,"name":"Kevin Carter","email":"kevin@cloudnull.com","username":"cloudnull"},"ref":"refs/changes/32/664932/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/32/664932/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/32/664932/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/32/664932/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/32/664932/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/32/664932/1"}}},"commit":{"parents":[{"commit":"4f5854701438f5af5f92734417bdaa803500ab91","subject":"Replace the old Stein cycle signing key with Train","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/4f5854701438f5af5f92734417bdaa803500ab91"}]}],"author":{"name":"Kevin Carter","email":"kecarter@redhat.com","date":"2019-06-12 13:52:57.000000000","tz":-300},"committer":{"name":"Kevin Carter","email":"kecarter@redhat.com","date":"2019-06-12 13:54:13.000000000","tz":-300},"subject":"Retire independent ansible role","message":"Retire independent ansible role\n\nAll of the tripleo ansible roles have been retired. The transformation squad\nis pivoting away from independent roles at this time and will be focusing on\nthe tripleo-ansible repository. For more information on the tripleo-ansible\nefforts and the direction of the transformation squad please review the\nfollowing mailing list thread [0].\n\n[0] - http://lists.openstack.org/pipermail/openstack-discuss/2019-June/006836.html\n\nChange-Id: Ia22150be91e7f45f59be4bab381727faa3d637df\nSigned-off-by: Kevin Carter \u003ckecarter@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/d038fec288638df9de5bd40d65bebe6aedd3aa25"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/d038fec288638df9de5bd40d65bebe6aedd3aa25"}]},"branch":"refs/heads/master"},"9716bcfd58f6ae875ed0b532bd22258a9fb1f4c3":{"kind":"REWORK","_number":2,"created":"2019-06-12 15:11:07.000000000","uploader":{"_account_id":7353,"name":"Kevin Carter","email":"kevin@cloudnull.com","username":"cloudnull"},"ref":"refs/changes/32/664932/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/32/664932/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/32/664932/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/32/664932/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/32/664932/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/32/664932/2"}}},"commit":{"parents":[{"commit":"4f5854701438f5af5f92734417bdaa803500ab91","subject":"Replace the old Stein cycle signing key with Train","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/4f5854701438f5af5f92734417bdaa803500ab91"}]}],"author":{"name":"Kevin Carter","email":"kecarter@redhat.com","date":"2019-06-12 13:52:57.000000000","tz":-300},"committer":{"name":"Kevin Carter (cloudnull)","email":"kevin@cloudnull.com","date":"2019-06-12 14:33:13.000000000","tz":0},"subject":"Retire independent ansible role","message":"Retire independent ansible role\n\nAll of the tripleo ansible roles have been retired. The transformation squad\nis pivoting away from independent roles at this time and will be focusing on\nthe tripleo-ansible repository. For more information on the tripleo-ansible\nefforts and the direction of the transformation squad please review the\nfollowing mailing list thread [0].\n\n[0] - http://lists.openstack.org/pipermail/openstack-discuss/2019-June/006836.html\n\nChange-Id: Ia22150be91e7f45f59be4bab381727faa3d637df\nSigned-off-by: Kevin Carter \u003ckecarter@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/9716bcfd58f6ae875ed0b532bd22258a9fb1f4c3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/9716bcfd58f6ae875ed0b532bd22258a9fb1f4c3"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
