)]}'
{"id":"openstack%2Fproject-config~665037","triplet_id":"openstack%2Fproject-config~master~I250596015d9d0fa4f37112c05c48a70957968ac1","project":"openstack/project-config","branch":"master","topic":"kafs-mirror","hashtags":[],"change_id":"I250596015d9d0fa4f37112c05c48a70957968ac1","subject":"Revert \"Revert \"opendev.org mirrors: switch out DFW, switch in IAD\"\"","status":"MERGED","created":"2019-06-12 23:43:12.000000000","updated":"2019-06-13 00:09:30.000000000","submitted":"2019-06-13 00:09:30.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"665037-1560384570369-216c3fa0","meta_rev_id":"dd2f66827a3c3dbe83e2d361b1ffaa2bfe7d4dee","_number":665037,"virtual_id_number":665037,"owner":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":1,"name":"James E. Blair","email":"jim@acmegating.com","username":"corvus"},{"value":0,"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"value":2,"date":"2019-06-13 00:09:29.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":1,"name":"James E. Blair","email":"jim@acmegating.com","username":"corvus"},"all":[{"value":2,"date":"2019-06-12 23:45:43.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":1,"name":"James E. Blair","email":"jim@acmegating.com","username":"corvus"},{"value":2,"date":"2019-06-12 23:43:44.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":1,"name":"James E. Blair","email":"jim@acmegating.com","username":"corvus"},"all":[{"value":1,"date":"2019-06-12 23:45:43.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":1,"name":"James E. Blair","email":"jim@acmegating.com","username":"corvus"},{"value":0,"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1,"name":"James E. Blair","email":"jim@acmegating.com","username":"corvus"},{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-06-12 23:43:44.000000000","updated_by":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"reviewer":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"state":"REVIEWER"},{"updated":"2019-06-12 23:45:43.000000000","updated_by":{"_account_id":1,"name":"James E. Blair","email":"jim@acmegating.com","username":"corvus"},"reviewer":{"_account_id":1,"name":"James E. Blair","email":"jim@acmegating.com","username":"corvus"},"state":"REVIEWER"},{"updated":"2019-06-13 00:09:29.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"47c46e67f207d4e760d95a06735db5cd67fc3b3e","author":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"date":"2019-06-12 23:43:12.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"c4c84049e808f0aa9b3b876b26c69da7daeca703","author":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"date":"2019-06-12 23:43:44.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"acf3b0074d332dfe14e950571bdd124d7c12df6a","author":{"_account_id":1,"name":"James E. Blair","email":"jim@acmegating.com","username":"corvus"},"date":"2019-06-12 23:45:43.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"c68774c863144d64de516a0f2db73eba5b5b90db","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-12 23:56:27.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-zuul-jobs-linters http://logs.openstack.org/37/665037/1/check/openstack-zuul-jobs-linters/8d5d41a/ : SUCCESS in 6m 30s","accounts_in_message":[],"_revision_number":1},{"id":"a587dd98b938fb14211800d2e2e6d1126f1fcf02","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-12 23:56:41.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"e5814912ee2fb10bc21f83547df3cb3d61cfaf73","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-13 00:09:29.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-zuul-jobs-linters http://logs.openstack.org/37/665037/1/gate/openstack-zuul-jobs-linters/0098698/ : SUCCESS in 7m 05s","accounts_in_message":[],"_revision_number":1},{"id":"2ea52d1b85bb583e5db07dbf01262fd1a3c4063b","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-13 00:09:30.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"415bf0a477dfb4cbca66858870314382dc4c8d42","revisions":{"415bf0a477dfb4cbca66858870314382dc4c8d42":{"kind":"REWORK","_number":1,"created":"2019-06-12 23:43:12.000000000","uploader":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"ref":"refs/changes/37/665037/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/37/665037/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/37/665037/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/37/665037/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/37/665037/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/37/665037/1"}}},"commit":{"parents":[{"commit":"d8b7f2da286f88a4bb687897f6f514384871aa6e","subject":"Revert \"opendev.org mirrors: switch out DFW, switch in IAD\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/d8b7f2da286f88a4bb687897f6f514384871aa6e"}]}],"author":{"name":"Ian Wienand","email":"iwienand@redhat.com","date":"2019-06-12 23:41:18.000000000","tz":600},"committer":{"name":"Ian Wienand","email":"iwienand@redhat.com","date":"2019-06-12 23:42:57.000000000","tz":600},"subject":"Revert \"Revert \"opendev.org mirrors: switch out DFW, switch in IAD\"\"","message":"Revert \"Revert \"opendev.org mirrors: switch out DFW, switch in IAD\"\"\n\nThis reverts commit d8b7f2da286f88a4bb687897f6f514384871aa6e.\n\nThe IAD server is now running 5.1.9 stable kernel, so we can try again\nfor a little and see what we get.  We also have some debugging plans\n(capturing /proc/\u003cpid\u003e/stack for apache) if it doesn\u0027t quite work.\n\nChange-Id: I250596015d9d0fa4f37112c05c48a70957968ac1\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/415bf0a477dfb4cbca66858870314382dc4c8d42"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/415bf0a477dfb4cbca66858870314382dc4c8d42"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
