)]}'
{"id":"openstack%2Fproject-config~665910","triplet_id":"openstack%2Fproject-config~master~I2afc533926e375d3b3523eec4e90798f367bb10b","project":"openstack/project-config","branch":"master","topic":"new-project","hashtags":[],"change_id":"I2afc533926e375d3b3523eec4e90798f367bb10b","subject":"Adding CI for ironic-prometheus-exporter","status":"MERGED","created":"2019-06-18 08:37:52.000000000","updated":"2019-07-09 07:32:26.000000000","submitted":"2019-07-09 07:32:26.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"665910-1562657546808-7de5f3e4","meta_rev_id":"85d510de3d8be78129b193e3c32e862428e93fd4","_number":665910,"virtual_id_number":665910,"owner":{"_account_id":15519,"name":"Iury Gregory Melo Ferreira","display_name":"Iury Gregory","email":"iurygregory@gmail.com","username":"iurygregory"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":0,"_account_id":6794,"name":"Simon Pasquier","email":"pasquier.simon@gmail.com","username":"simon-pasquier"},{"value":2,"date":"2019-07-09 07:32:26.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"all":[{"value":2,"date":"2019-07-05 14:02:14.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":0,"_account_id":6794,"name":"Simon Pasquier","email":"pasquier.simon@gmail.com","username":"simon-pasquier"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2019-07-09 07:22:16.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"all":[{"value":0,"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":0,"date":"2019-07-09 07:24:09.000000000","_account_id":6794,"name":"Simon Pasquier","email":"pasquier.simon@gmail.com","username":"simon-pasquier"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2019-07-09 07:22:16.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"_account_id":6794,"name":"Simon Pasquier","email":"pasquier.simon@gmail.com","username":"simon-pasquier"},{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-07-05 14:02:14.000000000","updated_by":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"reviewer":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"state":"REVIEWER"},{"updated":"2019-07-09 07:22:16.000000000","updated_by":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"reviewer":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"state":"REVIEWER"},{"updated":"2019-07-09 07:24:09.000000000","updated_by":{"_account_id":6794,"name":"Simon Pasquier","email":"pasquier.simon@gmail.com","username":"simon-pasquier"},"reviewer":{"_account_id":6794,"name":"Simon Pasquier","email":"pasquier.simon@gmail.com","username":"simon-pasquier"},"state":"REVIEWER"},{"updated":"2019-07-09 07:32:26.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"e358e8f57acbdac0eba9b2c487b51442d3932b6b","author":{"_account_id":15519,"name":"Iury Gregory Melo Ferreira","display_name":"Iury Gregory","email":"iurygregory@gmail.com","username":"iurygregory"},"date":"2019-06-18 08:37:52.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"b2e28fdf744417602e44aa341d7351002d5cc3f1","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-18 08:38:11.000000000","message":"Patch Set 1: Verified-1\n\n(1 comment)\n\nZuul encountered a syntax error while parsing its configuration in the\nrepo openstack/project-config on branch master.  The error was:\n\n  The project \"openstack/ironic-prometheus-exporter\" was not found.  All\n  projects referenced within a Zuul configuration must first be added to\n  the main configuration file by the Zuul administrator.\n\nThe error appears in the following project stanza:\n\n  project:\n      name: openstack/ironic-prometheus-exporter\n      templates:\n        - publish-to-pypi\n\n  in \"openstack/project-config/zuul.d/projects.yaml@master\", line 2628, column 3","accounts_in_message":[],"_revision_number":1},{"id":"f582b46807772414906e10148d31357e6f42dded","author":{"_account_id":15519,"name":"Iury Gregory Melo Ferreira","display_name":"Iury Gregory","email":"iurygregory@gmail.com","username":"iurygregory"},"date":"2019-06-20 13:54:39.000000000","message":"Patch Set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"2a478b3980a1e0642ae4680966cc4705e0aeba95","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-20 13:55:28.000000000","message":"Patch Set 2: Verified-1\n\n(1 comment)\n\nZuul encountered a syntax error while parsing its configuration in the\nrepo openstack/project-config on branch master.  The error was:\n\n  The project \"openstack/ironic-prometheus-exporter\" was not found.  All\n  projects referenced within a Zuul configuration must first be added to\n  the main configuration file by the Zuul administrator.\n\nThe error appears in the following project stanza:\n\n  project:\n      name: openstack/ironic-prometheus-exporter\n      templates:\n        - publish-to-pypi\n\n  in \"openstack/project-config/zuul.d/projects.yaml@master\", line 2628, column 3","accounts_in_message":[],"_revision_number":2},{"id":"f5697414f1a0eb45fc0841bcf5ee281cde4e9157","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2019-07-05 07:18:04.000000000","message":"Patch Set 2: Code-Review-1\n\nDependency is merged, please rebase to resolve merge conflict.","accounts_in_message":[],"_revision_number":2},{"id":"2b3754873c3fdf6e7ba46078ebdd903490477254","author":{"_account_id":15519,"name":"Iury Gregory Melo Ferreira","display_name":"Iury Gregory","email":"iurygregory@gmail.com","username":"iurygregory"},"date":"2019-07-05 13:42:46.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"633c97a8f370326713e6bb3d928ac316092bf8cf","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-05 13:49:56.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-zuul-jobs-linters http://logs.openstack.org/10/665910/3/check/openstack-zuul-jobs-linters/a334368/ : SUCCESS in 6m 31s","accounts_in_message":[],"_revision_number":3},{"id":"0aa64f3ea5d24744ec81c1ce9053c95958813e8b","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2019-07-05 14:02:14.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"4e835e5b49b5e338e925bfdced5708a704ba17cc","author":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"date":"2019-07-09 07:22:16.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"afe62a80ec1d443cf3964200a08c97f90247d1d3","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-09 07:22:34.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"cd097da587545c6d92f2007aa56e1d0cb4db1c08","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-09 07:32:26.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-zuul-jobs-linters http://logs.openstack.org/10/665910/3/gate/openstack-zuul-jobs-linters/daca226/ : SUCCESS in 7m 11s","accounts_in_message":[],"_revision_number":3},{"id":"22c86bbef14d83f780a5d971fa4001b3e41aa9e6","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-09 07:32:26.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"a9b44eb7729c90ae8b5c11dced324322fbcff650","revisions":{"cdcd2d769dc42b09a3ac2400762ad43032d14063":{"kind":"REWORK","_number":1,"created":"2019-06-18 08:37:52.000000000","uploader":{"_account_id":15519,"name":"Iury Gregory Melo Ferreira","display_name":"Iury Gregory","email":"iurygregory@gmail.com","username":"iurygregory"},"ref":"refs/changes/10/665910/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/10/665910/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/10/665910/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/10/665910/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/10/665910/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/10/665910/1"}}},"commit":{"parents":[{"commit":"533130debff525bd4708501a8d2056ae76ed2d7f","subject":"Merge \"Add airship/election\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/533130debff525bd4708501a8d2056ae76ed2d7f"}]}],"author":{"name":"Iury Gregory Melo Ferreira","email":"imelofer@redhat.com","date":"2019-06-18 08:35:07.000000000","tz":120},"committer":{"name":"Iury Gregory Melo Ferreira","email":"imelofer@redhat.com","date":"2019-06-18 08:36:05.000000000","tz":120},"subject":"Adding CI for ironic-prometheus-exporter","message":"Adding CI for ironic-prometheus-exporter\n\nRequired change for adding ironic-prometheus-exporter\nunder ironic umbrella I3fe01ec5465e4ddc1eb7966beca96823c2dbda4b\n\nChange-Id: I2afc533926e375d3b3523eec4e90798f367bb10b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/cdcd2d769dc42b09a3ac2400762ad43032d14063"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/cdcd2d769dc42b09a3ac2400762ad43032d14063"}]},"branch":"refs/heads/master"},"496e81476b017414770e3fc60d908d8c50efee02":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2019-06-20 13:54:39.000000000","uploader":{"_account_id":15519,"name":"Iury Gregory Melo Ferreira","display_name":"Iury Gregory","email":"iurygregory@gmail.com","username":"iurygregory"},"ref":"refs/changes/10/665910/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/10/665910/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/10/665910/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/10/665910/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/10/665910/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/10/665910/2"}}},"commit":{"parents":[{"commit":"533130debff525bd4708501a8d2056ae76ed2d7f","subject":"Merge \"Add airship/election\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/533130debff525bd4708501a8d2056ae76ed2d7f"}]}],"author":{"name":"Iury Gregory Melo Ferreira","email":"imelofer@redhat.com","date":"2019-06-18 08:35:07.000000000","tz":120},"committer":{"name":"Iury Gregory Melo Ferreira","email":"iurygregory@gmail.com","date":"2019-06-20 13:54:31.000000000","tz":0},"subject":"Adding CI for ironic-prometheus-exporter","message":"Adding CI for ironic-prometheus-exporter\n\nRequired change for adding ironic-prometheus-exporter\nunder ironic umbrella I3fe01ec5465e4ddc1eb7966beca96823c2dbda4b\n\nDepends-On: I3fe01ec5465e4ddc1eb7966beca96823c2dbda4b\nChange-Id: I2afc533926e375d3b3523eec4e90798f367bb10b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/496e81476b017414770e3fc60d908d8c50efee02"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/496e81476b017414770e3fc60d908d8c50efee02"}]},"branch":"refs/heads/master"},"a9b44eb7729c90ae8b5c11dced324322fbcff650":{"kind":"REWORK","_number":3,"created":"2019-07-05 13:42:46.000000000","uploader":{"_account_id":15519,"name":"Iury Gregory Melo Ferreira","display_name":"Iury Gregory","email":"iurygregory@gmail.com","username":"iurygregory"},"ref":"refs/changes/10/665910/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/10/665910/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/10/665910/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/10/665910/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/10/665910/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/10/665910/3"}}},"commit":{"parents":[{"commit":"69b6379e677a3e11dadbff1795c3796f015f4395","subject":"Move gerritbot to #openstack-kolla for kayobe","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/69b6379e677a3e11dadbff1795c3796f015f4395"}]}],"author":{"name":"Iury Gregory Melo Ferreira","email":"imelofer@redhat.com","date":"2019-06-18 08:35:07.000000000","tz":120},"committer":{"name":"Iury Gregory Melo Ferreira","email":"imelofer@redhat.com","date":"2019-07-05 13:41:28.000000000","tz":120},"subject":"Adding CI for ironic-prometheus-exporter","message":"Adding CI for ironic-prometheus-exporter\n\nRequired change for adding ironic-prometheus-exporter\nunder ironic umbrella I3fe01ec5465e4ddc1eb7966beca96823c2dbda4b\n\nDepends-On: I3fe01ec5465e4ddc1eb7966beca96823c2dbda4b\nChange-Id: I2afc533926e375d3b3523eec4e90798f367bb10b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/a9b44eb7729c90ae8b5c11dced324322fbcff650"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/a9b44eb7729c90ae8b5c11dced324322fbcff650"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
