)]}'
{"id":"openstack%2Fproject-config~667418","triplet_id":"openstack%2Fproject-config~master~Id0842ec0a838246b701cb3e04b69f9b4727fdcef","project":"openstack/project-config","branch":"master","topic":"retire-role","hashtags":[],"change_id":"Id0842ec0a838246b701cb3e04b69f9b4727fdcef","subject":"Remove retired repos","status":"MERGED","created":"2019-06-25 17:21:49.000000000","updated":"2019-06-26 17:55:44.000000000","submitted":"2019-06-26 17:55:44.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":4,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"667418-1561571744994-c8b2cb04","meta_rev_id":"6b69c6f7b56132232a3e64f8ade62f69d9b8c652","_number":667418,"virtual_id_number":667418,"owner":{"_account_id":7353,"name":"Kevin Carter","email":"kevin@cloudnull.com","username":"cloudnull"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":1,"name":"James E. Blair","email":"jim@acmegating.com","username":"corvus"},{"value":0,"_account_id":24162,"name":"Sorin Sbârnea","display_name":"zbr","email":"ssbarnea@redhat.com","username":"ssbarnea","status":"do not feed the troll 🥕"},{"value":0,"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"value":0,"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":0,"_account_id":7353,"name":"Kevin Carter","email":"kevin@cloudnull.com","username":"cloudnull"},{"value":2,"date":"2019-06-26 17:55:44.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":1,"name":"James E. Blair","email":"jim@acmegating.com","username":"corvus"},"all":[{"value":2,"date":"2019-06-26 17:43:33.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":1,"name":"James E. Blair","email":"jim@acmegating.com","username":"corvus"},{"value":1,"date":"2019-06-25 19:38:38.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":24162,"name":"Sorin Sbârnea","display_name":"zbr","email":"ssbarnea@redhat.com","username":"ssbarnea","status":"do not feed the troll 🥕"},{"value":0,"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"value":2,"date":"2019-06-26 17:36:59.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":0,"_account_id":7353,"name":"Kevin Carter","email":"kevin@cloudnull.com","username":"cloudnull"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":1,"name":"James E. Blair","email":"jim@acmegating.com","username":"corvus"},"all":[{"value":1,"date":"2019-06-26 17:43:33.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":1,"name":"James E. Blair","email":"jim@acmegating.com","username":"corvus"},{"value":0,"_account_id":24162,"name":"Sorin Sbârnea","display_name":"zbr","email":"ssbarnea@redhat.com","username":"ssbarnea","status":"do not feed the troll 🥕"},{"value":0,"date":"2019-06-26 13:31:57.000000000","permitted_voting_range":{"min":0,"max":1},"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"value":0,"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":0,"_account_id":7353,"name":"Kevin Carter","email":"kevin@cloudnull.com","username":"cloudnull"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"date":"2019-06-26 13:32:01.000000000","permitted_voting_range":{"min":0,"max":1},"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1,"name":"James E. Blair","email":"jim@acmegating.com","username":"corvus"},{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"_account_id":7353,"name":"Kevin Carter","email":"kevin@cloudnull.com","username":"cloudnull"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":24162,"name":"Sorin Sbârnea","display_name":"zbr","email":"ssbarnea@redhat.com","username":"ssbarnea","status":"do not feed the troll 🥕"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-06-25 19:38:38.000000000","updated_by":{"_account_id":24162,"name":"Sorin Sbârnea","display_name":"zbr","email":"ssbarnea@redhat.com","username":"ssbarnea","status":"do not feed the troll 🥕"},"reviewer":{"_account_id":24162,"name":"Sorin Sbârnea","display_name":"zbr","email":"ssbarnea@redhat.com","username":"ssbarnea","status":"do not feed the troll 🥕"},"state":"REVIEWER"},{"updated":"2019-06-26 13:31:57.000000000","updated_by":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"reviewer":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"state":"REVIEWER"},{"updated":"2019-06-26 13:32:01.000000000","updated_by":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"reviewer":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"state":"REVIEWER"},{"updated":"2019-06-26 17:36:59.000000000","updated_by":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"reviewer":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"state":"REVIEWER"},{"updated":"2019-06-26 17:43:33.000000000","updated_by":{"_account_id":1,"name":"James E. Blair","email":"jim@acmegating.com","username":"corvus"},"reviewer":{"_account_id":1,"name":"James E. Blair","email":"jim@acmegating.com","username":"corvus"},"state":"REVIEWER"},{"updated":"2019-06-26 17:55:44.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"eacd49d69ef15c7e87c90cb3f2d61934192d8ceb","author":{"_account_id":7353,"name":"Kevin Carter","email":"kevin@cloudnull.com","username":"cloudnull"},"date":"2019-06-25 17:21:49.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"243056372e993a7a5eaa91d1d8429b9d4320116d","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2019-06-25 17:24:46.000000000","message":"Patch Set 1: Code-Review-1\n\nThis is not enough for full retirement:\n\n* Remove from zuul/main.yaml\n* Update gerrit/projects.yaml\n* remove gerrit/acl/ files if exists\n\nSee how similar changes are done","accounts_in_message":[],"_revision_number":1},{"id":"9f1f074130d020cac6025406f0a0925140a19ff7","author":{"_account_id":7353,"name":"Kevin Carter","email":"kevin@cloudnull.com","username":"cloudnull"},"date":"2019-06-25 17:35:14.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"4fa4608e8d71d7c060d3e814ea15dde5f7ce7910","author":{"_account_id":7353,"name":"Kevin Carter","email":"kevin@cloudnull.com","username":"cloudnull"},"date":"2019-06-25 17:36:38.000000000","message":"Patch Set 2:\n\n\u003e This is not enough for full retirement:\n \u003e \n \u003e * Remove from zuul/main.yaml\n \u003e * Update gerrit/projects.yaml\n \u003e * remove gerrit/acl/ files if exists\n \u003e \n \u003e See how similar changes are done\n\nI didn\u0027t see any acl\u0027s that need to be removed at this time; the one role we\u0027re keeping is `ansible-role-tripleo-modify-image`.","accounts_in_message":[],"_revision_number":2},{"id":"221a750b6d063daa82907d3c0b553275ed7e35b7","author":{"_account_id":7353,"name":"Kevin Carter","email":"kevin@cloudnull.com","username":"cloudnull"},"date":"2019-06-25 17:39:51.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"89192c903b1ff3301a382ec06f5c5fd8a97cebe7","author":{"_account_id":7353,"name":"Kevin Carter","email":"kevin@cloudnull.com","username":"cloudnull"},"date":"2019-06-25 17:41:23.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"3fe5e8dfdb646d536c4c28d5bd94c58474238d37","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2019-06-25 17:47:47.000000000","message":"Patch Set 4: Code-Review-1\n\n(2 comments)\n\nplease remove all the groups entries...","accounts_in_message":[],"_revision_number":4},{"id":"54bf20b630ff6a48bec93f70a97bc6eddc29498b","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-25 17:53:46.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-zuul-jobs-linters http://logs.openstack.org/18/667418/4/check/openstack-zuul-jobs-linters/29b7aa2/ : SUCCESS in 6m 58s\n- project-config-gerrit http://logs.openstack.org/18/667418/4/check/project-config-gerrit/8c3d8fa/ : SUCCESS in 4m 39s","accounts_in_message":[],"_revision_number":4},{"id":"918ef4e7d18668fffda10d28fae4774945828126","author":{"_account_id":7353,"name":"Kevin Carter","email":"kevin@cloudnull.com","username":"cloudnull"},"date":"2019-06-25 18:40:48.000000000","message":"Patch Set 4:\n\n(2 comments)","accounts_in_message":[],"_revision_number":4},{"id":"f4e89f29415c1401c3aa7f4928be1d2122b97899","author":{"_account_id":7353,"name":"Kevin Carter","email":"kevin@cloudnull.com","username":"cloudnull"},"date":"2019-06-25 18:41:09.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"ad643e59fab2da5b25f37851a75b5f4e7effe6b1","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2019-06-25 18:44:59.000000000","message":"Patch Set 5: Code-Review+2","accounts_in_message":[],"_revision_number":5},{"id":"3846918c3e078fc383928e00ee9ebc92031790f6","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-25 18:53:24.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-zuul-jobs-linters http://logs.openstack.org/18/667418/5/check/openstack-zuul-jobs-linters/d666581/ : SUCCESS in 6m 12s\n- project-config-gerrit http://logs.openstack.org/18/667418/5/check/project-config-gerrit/0c95ab2/ : SUCCESS in 3m 23s","accounts_in_message":[],"_revision_number":5},{"id":"2145f3aa3abb722cf1cc960e5c5a216e226068a9","author":{"_account_id":24162,"name":"Sorin Sbârnea","display_name":"zbr","email":"ssbarnea@redhat.com","username":"ssbarnea","status":"do not feed the troll 🥕"},"date":"2019-06-25 19:38:38.000000000","message":"Patch Set 5: Code-Review+1\n\nLGTM, in fact very good.","accounts_in_message":[],"_revision_number":5},{"id":"4ac9e21be2ee7df17048a91a33c361ed848efc76","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2019-06-26 14:12:49.000000000","message":"Patch Set 5: -Code-Review\n\nDo you have a followup change up for the governance repo to document the retirement?","accounts_in_message":[],"_revision_number":5},{"id":"16746c0dd3e5e54afe0d35cb4c2bd4d5a3b64935","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2019-06-26 17:36:59.000000000","message":"Patch Set 5: Code-Review+2\n\nhttps://review.opendev.org/#/c/667416/ is the governance change","accounts_in_message":[],"_revision_number":5},{"id":"6807ff1f3d8e7ed75b7258e64562f821a3573bee","author":{"_account_id":1,"name":"James E. Blair","email":"jim@acmegating.com","username":"corvus"},"date":"2019-06-26 17:43:33.000000000","message":"Patch Set 5: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":5},{"id":"7a34d3555790a5be6e5c668ef5a01d8663ad5ab0","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-26 17:43:53.000000000","message":"Patch Set 5: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":5},{"id":"4555ed5c10b1c96478b03bfd86bdcecb630b0982","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-26 17:55:44.000000000","message":"Patch Set 5: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-zuul-jobs-linters http://logs.openstack.org/18/667418/5/gate/openstack-zuul-jobs-linters/35b62dc/ : SUCCESS in 6m 10s\n- project-config-gerrit http://logs.openstack.org/18/667418/5/gate/project-config-gerrit/1d2a0dc/ : SUCCESS in 3m 46s","accounts_in_message":[],"_revision_number":5},{"id":"1b0fc63e9d833c8a6e63adb9e643e817409a625d","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-26 17:55:44.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":5}],"current_revision_number":5,"current_revision":"d7d46fc12cc4b6e48728707d8363979d9108e3de","revisions":{"70e04a8e72cd0edf93dd06412916db71b6fefa72":{"kind":"REWORK","_number":1,"created":"2019-06-25 17:21:49.000000000","uploader":{"_account_id":7353,"name":"Kevin Carter","email":"kevin@cloudnull.com","username":"cloudnull"},"ref":"refs/changes/18/667418/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/18/667418/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/18/667418/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/18/667418/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/18/667418/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/18/667418/1"}}},"commit":{"parents":[{"commit":"0f7bf508b2ccfe3a458bbc35dfdb3f0dc8ab189c","subject":"Fix retirement of ansible repos","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/0f7bf508b2ccfe3a458bbc35dfdb3f0dc8ab189c"}]}],"author":{"name":"Kevin Carter","email":"kecarter@redhat.com","date":"2019-06-25 17:21:11.000000000","tz":-300},"committer":{"name":"Kevin Carter","email":"kecarter@redhat.com","date":"2019-06-25 17:21:11.000000000","tz":-300},"subject":"Remove retired repos","message":"Remove retired repos\n\nThis removes the tripleo repos that have been retired from project config.\n\nChange-Id: Id0842ec0a838246b701cb3e04b69f9b4727fdcef\nSigned-off-by: Kevin Carter \u003ckecarter@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/70e04a8e72cd0edf93dd06412916db71b6fefa72"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/70e04a8e72cd0edf93dd06412916db71b6fefa72"}]},"branch":"refs/heads/master"},"03424efcbf2978fb6edfe12321d40541188f31ce":{"kind":"REWORK","_number":2,"created":"2019-06-25 17:35:14.000000000","uploader":{"_account_id":7353,"name":"Kevin Carter","email":"kevin@cloudnull.com","username":"cloudnull"},"ref":"refs/changes/18/667418/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/18/667418/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/18/667418/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/18/667418/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/18/667418/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/18/667418/2"}}},"commit":{"parents":[{"commit":"0f7bf508b2ccfe3a458bbc35dfdb3f0dc8ab189c","subject":"Fix retirement of ansible repos","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/0f7bf508b2ccfe3a458bbc35dfdb3f0dc8ab189c"}]}],"author":{"name":"Kevin Carter","email":"kecarter@redhat.com","date":"2019-06-25 17:21:11.000000000","tz":-300},"committer":{"name":"Kevin Carter","email":"kecarter@redhat.com","date":"2019-06-25 17:35:05.000000000","tz":-300},"subject":"Remove retired repos","message":"Remove retired repos\n\nThis removes the tripleo repos that have been retired from project config.\n\nChange-Id: Id0842ec0a838246b701cb3e04b69f9b4727fdcef\nSigned-off-by: Kevin Carter \u003ckecarter@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/03424efcbf2978fb6edfe12321d40541188f31ce"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/03424efcbf2978fb6edfe12321d40541188f31ce"}]},"branch":"refs/heads/master"},"958306098897be81183292da517132d80d43abc0":{"kind":"REWORK","_number":3,"created":"2019-06-25 17:39:51.000000000","uploader":{"_account_id":7353,"name":"Kevin Carter","email":"kevin@cloudnull.com","username":"cloudnull"},"ref":"refs/changes/18/667418/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/18/667418/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/18/667418/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/18/667418/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/18/667418/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/18/667418/3"}}},"commit":{"parents":[{"commit":"0f7bf508b2ccfe3a458bbc35dfdb3f0dc8ab189c","subject":"Fix retirement of ansible repos","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/0f7bf508b2ccfe3a458bbc35dfdb3f0dc8ab189c"}]}],"author":{"name":"Kevin Carter","email":"kecarter@redhat.com","date":"2019-06-25 17:21:11.000000000","tz":-300},"committer":{"name":"Kevin Carter","email":"kecarter@redhat.com","date":"2019-06-25 17:39:35.000000000","tz":-300},"subject":"Remove retired repos","message":"Remove retired repos\n\nThis removes the tripleo repos that have been retired from project config.\n\nChange-Id: Id0842ec0a838246b701cb3e04b69f9b4727fdcef\nSigned-off-by: Kevin Carter \u003ckecarter@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/958306098897be81183292da517132d80d43abc0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/958306098897be81183292da517132d80d43abc0"}]},"branch":"refs/heads/master"},"8ec81f3e74ef8eb30dcf2e1835687c8ef09c8cca":{"kind":"REWORK","_number":4,"created":"2019-06-25 17:41:23.000000000","uploader":{"_account_id":7353,"name":"Kevin Carter","email":"kevin@cloudnull.com","username":"cloudnull"},"ref":"refs/changes/18/667418/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/18/667418/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/18/667418/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/18/667418/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/18/667418/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/18/667418/4"}}},"commit":{"parents":[{"commit":"0f7bf508b2ccfe3a458bbc35dfdb3f0dc8ab189c","subject":"Fix retirement of ansible repos","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/0f7bf508b2ccfe3a458bbc35dfdb3f0dc8ab189c"}]}],"author":{"name":"Kevin Carter","email":"kecarter@redhat.com","date":"2019-06-25 17:21:11.000000000","tz":-300},"committer":{"name":"Kevin Carter","email":"kecarter@redhat.com","date":"2019-06-25 17:40:37.000000000","tz":-300},"subject":"Remove retired repos","message":"Remove retired repos\n\nThis removes the tripleo repos that have been retired from project config.\n\nChange-Id: Id0842ec0a838246b701cb3e04b69f9b4727fdcef\nSigned-off-by: Kevin Carter \u003ckecarter@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/8ec81f3e74ef8eb30dcf2e1835687c8ef09c8cca"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/8ec81f3e74ef8eb30dcf2e1835687c8ef09c8cca"}]},"branch":"refs/heads/master"},"d7d46fc12cc4b6e48728707d8363979d9108e3de":{"kind":"REWORK","_number":5,"created":"2019-06-25 18:41:09.000000000","uploader":{"_account_id":7353,"name":"Kevin Carter","email":"kevin@cloudnull.com","username":"cloudnull"},"ref":"refs/changes/18/667418/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/18/667418/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/18/667418/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/18/667418/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/18/667418/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/18/667418/5"}}},"commit":{"parents":[{"commit":"0f7bf508b2ccfe3a458bbc35dfdb3f0dc8ab189c","subject":"Fix retirement of ansible repos","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/0f7bf508b2ccfe3a458bbc35dfdb3f0dc8ab189c"}]}],"author":{"name":"Kevin Carter","email":"kecarter@redhat.com","date":"2019-06-25 17:21:11.000000000","tz":-300},"committer":{"name":"Kevin Carter","email":"kecarter@redhat.com","date":"2019-06-25 18:41:02.000000000","tz":-300},"subject":"Remove retired repos","message":"Remove retired repos\n\nThis removes the tripleo repos that have been retired from project config.\n\nChange-Id: Id0842ec0a838246b701cb3e04b69f9b4727fdcef\nSigned-off-by: Kevin Carter \u003ckecarter@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/d7d46fc12cc4b6e48728707d8363979d9108e3de"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/d7d46fc12cc4b6e48728707d8363979d9108e3de"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
