)]}'
{"id":"openstack%2Fproject-config~667949","triplet_id":"openstack%2Fproject-config~master~I35c7446f2c52c41b7e4652af747e7af445d71609","project":"openstack/project-config","branch":"master","topic":"retire-tempest-tripleo-ui","hashtags":[],"change_id":"I35c7446f2c52c41b7e4652af747e7af445d71609","subject":"Retire tempest-tripleo-ui","status":"MERGED","created":"2019-06-27 14:50:39.000000000","updated":"2019-07-09 07:30:36.000000000","submitted":"2019-07-09 07:30:35.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"667949-1562657436186-0a1ee91c","meta_rev_id":"219a73d44081c2225083e486ac9026633850dd36","_number":667949,"virtual_id_number":667949,"owner":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":2,"date":"2019-07-09 07:30:35.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"all":[{"value":2,"date":"2019-07-08 17:14:47.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2019-07-09 07:21:44.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"all":[{"value":0,"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2019-07-09 07:21:44.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-07-08 17:14:47.000000000","updated_by":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"reviewer":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"state":"REVIEWER"},{"updated":"2019-07-09 07:21:44.000000000","updated_by":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"reviewer":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"state":"REVIEWER"},{"updated":"2019-07-09 07:30:35.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"a1b22b4ae3aa84ee4306e0baf9721a31daf14d66","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2019-06-27 14:50:39.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"75835cac48b44a05536702cea876a682d744460e","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-27 15:03:27.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-zuul-jobs-linters http://logs.openstack.org/49/667949/1/check/openstack-zuul-jobs-linters/0722e36/ : FAILURE in 3m 48s","accounts_in_message":[],"_revision_number":1},{"id":"e94a16d56e753476c023bb8267699e5e4d556d2d","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2019-06-27 15:12:10.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"f8bdf00c875aeff8ca85cedb0206585ff8cf2094","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2019-06-27 15:18:50.000000000","message":"Patch Set 2:\n\nneeded-by is wrong - you need a needed-by pointing to a change that removes the content on master branch....","accounts_in_message":[],"_revision_number":2},{"id":"205d5d1629f329c3fdb6ba106e88e8609b010723","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2019-06-27 15:19:51.000000000","message":"Patch Set 3: Commit message was updated.","accounts_in_message":[],"_revision_number":3},{"id":"7044c27f0ad5f1943a225877fd99f225e490d717","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2019-06-27 15:27:17.000000000","message":"Patch Set 3: Code-Review-1\n\nthe repo has only a master branch - so, you retire \"all\" existing branches and the master lines are not needed.\n\nDon\u0027t you need master of tempest-tripleo-ui for stable tripleo-ui work?","accounts_in_message":[],"_revision_number":3},{"id":"601de09fc2648a3dcdd472255addce898eebc233","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2019-06-27 15:29:47.000000000","message":"Patch Set 3:\n\nOh - you already have ACLs set to retired for tempest-tripleo-ui, so you need to enable those first if you need to merge another change there. Note that this is involves a manual infra-root action.","accounts_in_message":[],"_revision_number":3},{"id":"ef803deca6179d26f25406938fc299d9b3d131b8","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2019-06-27 15:30:35.000000000","message":"Patch Set 3:\n\nNote also that tempest-tripleo-ui is mentioned in various places, better remove those first: http://codesearch.openstack.org/?q\u003dtempest-tripleo-ui\u0026i\u003dnope\u0026files\u003d\u0026repos\u003d","accounts_in_message":[],"_revision_number":3},{"id":"a8e8d3b99f7f8034cb8724b571853a20f4ea88fa","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-27 15:34:00.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-zuul-jobs-linters http://logs.openstack.org/49/667949/3/check/openstack-zuul-jobs-linters/e58870e/ : SUCCESS in 6m 58s","accounts_in_message":[],"_revision_number":3},{"id":"5c1c3ab9abfa2ad8e4d43a83b858a03f1862a797","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2019-06-27 15:38:37.000000000","message":"Patch Set 3:\n\ntempest-tripleo-ui was never fully developed. I\u0027ll figure out the cleanup actions","accounts_in_message":[],"_revision_number":3},{"id":"0042af38b93c6466d20d75ec8f81c92c9acda559","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2019-07-08 17:14:03.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"a95a3de5e34fd92fc18f1d773c204b3c19e9dfa5","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2019-07-08 17:14:47.000000000","message":"Patch Set 4: Code-Review+2","accounts_in_message":[],"_revision_number":4},{"id":"ed8134b577275e97a7ab954983fad6972e685576","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-08 17:26:48.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-zuul-jobs-linters http://logs.openstack.org/49/667949/4/check/openstack-zuul-jobs-linters/b338486/ : SUCCESS in 6m 41s","accounts_in_message":[],"_revision_number":4},{"id":"ba05519f5c0e5cc89be068daf82ab4e73fccd6d4","author":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"date":"2019-07-09 07:21:44.000000000","message":"Patch Set 4: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":4},{"id":"48b3804d7f4c5178162ac4d4cd603026a499925c","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-09 07:22:02.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"199af30812862219eda5fca20569ae91d57058a4","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-09 07:30:35.000000000","message":"Patch Set 4: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-zuul-jobs-linters http://logs.openstack.org/49/667949/4/gate/openstack-zuul-jobs-linters/87bf337/ : SUCCESS in 6m 55s","accounts_in_message":[],"_revision_number":4},{"id":"0cf5ffd9ab0d486bc586eec3c302beb89e861b1a","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-09 07:30:36.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"ac095748b90a79123f4de919b32de5d80b3a6081","revisions":{"d8f4b9707ee1468304a55cebd8990384525d406e":{"kind":"REWORK","_number":1,"created":"2019-06-27 14:50:39.000000000","uploader":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"ref":"refs/changes/49/667949/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/49/667949/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/49/667949/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/49/667949/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/49/667949/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/49/667949/1"}}},"commit":{"parents":[{"commit":"95592958c4d1761719f493ff04571315eb3da357","subject":"Merge \"Remove retired repos\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/95592958c4d1761719f493ff04571315eb3da357"}]}],"author":{"name":"Alex Schultz","email":"aschultz@redhat.com","date":"2019-06-27 14:49:17.000000000","tz":-360},"committer":{"name":"Alex Schultz","email":"aschultz@redhat.com","date":"2019-06-27 14:49:17.000000000","tz":-360},"subject":"Retire tempest-tripleo-ui","message":"Retire tempest-tripleo-ui\n\nNeed to set to noop so we can remove content.\n\nChange-Id: I35c7446f2c52c41b7e4652af747e7af445d71609\nNeeded-By: https://review.opendev.org/#/c/665264/\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/d8f4b9707ee1468304a55cebd8990384525d406e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/d8f4b9707ee1468304a55cebd8990384525d406e"}]},"branch":"refs/heads/master"},"9d4a53a6f5654720ebcbfcd53c218bce741efef5":{"kind":"REWORK","_number":2,"created":"2019-06-27 15:12:10.000000000","uploader":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"ref":"refs/changes/49/667949/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/49/667949/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/49/667949/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/49/667949/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/49/667949/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/49/667949/2"}}},"commit":{"parents":[{"commit":"95592958c4d1761719f493ff04571315eb3da357","subject":"Merge \"Remove retired repos\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/95592958c4d1761719f493ff04571315eb3da357"}]}],"author":{"name":"Alex Schultz","email":"aschultz@redhat.com","date":"2019-06-27 14:49:17.000000000","tz":-360},"committer":{"name":"Alex Schultz","email":"aschultz@redhat.com","date":"2019-06-27 15:12:06.000000000","tz":-360},"subject":"Retire tempest-tripleo-ui","message":"Retire tempest-tripleo-ui\n\nNeed to set to noop so we can remove content.\n\nChange-Id: I35c7446f2c52c41b7e4652af747e7af445d71609\nNeeded-By: https://review.opendev.org/#/c/665264/\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/9d4a53a6f5654720ebcbfcd53c218bce741efef5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/9d4a53a6f5654720ebcbfcd53c218bce741efef5"}]},"branch":"refs/heads/master"},"ec0368c77a63b6e296cbbdacf666a7e17ca1e148":{"kind":"NO_CODE_CHANGE","_number":3,"created":"2019-06-27 15:19:51.000000000","uploader":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"ref":"refs/changes/49/667949/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/49/667949/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/49/667949/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/49/667949/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/49/667949/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/49/667949/3"}}},"commit":{"parents":[{"commit":"95592958c4d1761719f493ff04571315eb3da357","subject":"Merge \"Remove retired repos\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/95592958c4d1761719f493ff04571315eb3da357"}]}],"author":{"name":"Alex Schultz","email":"aschultz@redhat.com","date":"2019-06-27 14:49:17.000000000","tz":-360},"committer":{"name":"Alex Schultz","email":"aschultz@redhat.com","date":"2019-06-27 15:19:49.000000000","tz":0},"subject":"Retire tempest-tripleo-ui","message":"Retire tempest-tripleo-ui\n\nNeed to set to noop so we can remove content.\n\nChange-Id: I35c7446f2c52c41b7e4652af747e7af445d71609\nNeeded-By: https://review.opendev.org/#/c/667951/\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/ec0368c77a63b6e296cbbdacf666a7e17ca1e148"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/ec0368c77a63b6e296cbbdacf666a7e17ca1e148"}]},"branch":"refs/heads/master"},"ac095748b90a79123f4de919b32de5d80b3a6081":{"kind":"REWORK","_number":4,"created":"2019-07-08 17:14:03.000000000","uploader":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"ref":"refs/changes/49/667949/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/49/667949/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/49/667949/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/49/667949/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/49/667949/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/49/667949/4"}}},"commit":{"parents":[{"commit":"95592958c4d1761719f493ff04571315eb3da357","subject":"Merge \"Remove retired repos\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/95592958c4d1761719f493ff04571315eb3da357"}]}],"author":{"name":"Alex Schultz","email":"aschultz@redhat.com","date":"2019-06-27 14:49:17.000000000","tz":-360},"committer":{"name":"Alex Schultz","email":"aschultz@redhat.com","date":"2019-07-08 17:13:58.000000000","tz":-360},"subject":"Retire tempest-tripleo-ui","message":"Retire tempest-tripleo-ui\n\nNeed to set to noop so we can remove content.\n\nChange-Id: I35c7446f2c52c41b7e4652af747e7af445d71609\nNeeded-By: https://review.opendev.org/#/c/665264/\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/ac095748b90a79123f4de919b32de5d80b3a6081"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/ac095748b90a79123f4de919b32de5d80b3a6081"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
