)]}'
{"id":"openstack%2Fproject-config~668361","triplet_id":"openstack%2Fproject-config~master~I83c73a8cff9728336b6208cab04e7c97c4c4ad86","project":"openstack/project-config","branch":"master","topic":"gra1-kafs","hashtags":[],"change_id":"I83c73a8cff9728336b6208cab04e7c97c4c4ad86","subject":"Switch ovh/GRA1 to opendev.org mirror","status":"MERGED","created":"2019-07-01 06:57:07.000000000","updated":"2019-07-01 17:30:33.000000000","submitted":"2019-07-01 17:30:33.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"668361-1562002233806-887ddea0","meta_rev_id":"0ef6bc4d4f8b623e53f20656e643106cfd058651","_number":668361,"virtual_id_number":668361,"owner":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":2,"date":"2019-07-01 17:30:33.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},{"value":0,"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"all":[{"value":2,"date":"2019-07-01 07:58:42.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},{"value":2,"date":"2019-07-01 12:56:29.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"all":[{"value":0,"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},{"value":1,"date":"2019-07-01 12:56:29.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-07-01 07:58:42.000000000","updated_by":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"reviewer":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"state":"REVIEWER"},{"updated":"2019-07-01 12:56:29.000000000","updated_by":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"reviewer":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"state":"REVIEWER"},{"updated":"2019-07-01 17:30:33.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"3494bd4033f3fe0a3e2a3b63f1ff3db099d6646a","author":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"date":"2019-07-01 06:57:07.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"6910cbe1ec75d2c3618224bbb0d64f22fe43c57d","author":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"date":"2019-07-01 06:57:31.000000000","message":"Patch Set 1:\n\nNOTE(ianw): this should be monitored ...","accounts_in_message":[],"_revision_number":1},{"id":"27861b1efe0317b6966d055433c42a8239cc9138","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-01 07:08:51.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-zuul-jobs-linters http://logs.openstack.org/61/668361/1/check/openstack-zuul-jobs-linters/58d7db7/ : SUCCESS in 6m 44s","accounts_in_message":[],"_revision_number":1},{"id":"ac652eee34814c2f38e5f8bc98de9175bb85e52d","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2019-07-01 07:09:17.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"af50aa90561ba0c43385ca300a10c2247e075ab8","author":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"date":"2019-07-01 07:50:05.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"716e3818d79d0774e1b9443afad5ecccfc5def47","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2019-07-01 07:58:42.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"b4d99a33396e16fa71e17fa0ae7f9a4f5fcb49cf","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-01 08:03:20.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-zuul-jobs-linters http://logs.openstack.org/61/668361/2/check/openstack-zuul-jobs-linters/549cb59/ : SUCCESS in 6m 35s","accounts_in_message":[],"_revision_number":2},{"id":"f1fa9efb9c360bb41cce09b20859b4423bda3dae","author":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"date":"2019-07-01 12:56:29.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"bfcc8960ad4e27c717ed31d44df8b242cb6c1335","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-01 17:18:13.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"67706dd1fabf3e999c92b680ba2ddefbb40ffae5","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-01 17:30:33.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"4909ac0a2ae78cedba119d3a907bf53db778f83b","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-01 17:30:33.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-zuul-jobs-linters http://logs.openstack.org/61/668361/2/gate/openstack-zuul-jobs-linters/b4997ea/ : SUCCESS in 6m 43s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"45f3b85860123129630b612ef1f28c5456875892","revisions":{"903d7626201f877c6ceadf32313c26f37411fbca":{"kind":"REWORK","_number":1,"created":"2019-07-01 06:57:07.000000000","uploader":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"ref":"refs/changes/61/668361/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/61/668361/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/61/668361/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/61/668361/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/61/668361/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/61/668361/1"}}},"commit":{"parents":[{"commit":"c5c2e431daa2b48dfcdb6b22886666066f2aec61","subject":"Merge \"Removed unused ACL\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/c5c2e431daa2b48dfcdb6b22886666066f2aec61"}]}],"author":{"name":"Ian Wienand","email":"iwienand@redhat.com","date":"2019-07-01 06:50:49.000000000","tz":600},"committer":{"name":"Ian Wienand","email":"iwienand@redhat.com","date":"2019-07-01 06:50:49.000000000","tz":600},"subject":"Switch ovh/GRA1 to opendev.org mirror","message":"Switch ovh/GRA1 to opendev.org mirror\n\nThis is our second kafs based mirror (rax.IAD being the first), chosen\nto be geographically remote to the AFS servers.  It is currently\nmanually configured with 5.2-rc7 kernel, which has all the current\nkAFS fixes incorporated.\n\nSwitch it in for initial testing\n\nChange-Id: I83c73a8cff9728336b6208cab04e7c97c4c4ad86\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/903d7626201f877c6ceadf32313c26f37411fbca"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/903d7626201f877c6ceadf32313c26f37411fbca"}]},"branch":"refs/heads/master"},"45f3b85860123129630b612ef1f28c5456875892":{"kind":"REWORK","_number":2,"created":"2019-07-01 07:50:05.000000000","uploader":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"ref":"refs/changes/61/668361/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/61/668361/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/61/668361/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/61/668361/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/61/668361/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/61/668361/2"}}},"commit":{"parents":[{"commit":"0e0dc2e8fd6ad77b03e34a3d8fc15231d8330ea3","subject":"Switch RAX ORD mirrors to new opendev.org mirrors","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/0e0dc2e8fd6ad77b03e34a3d8fc15231d8330ea3"}]}],"author":{"name":"Ian Wienand","email":"iwienand@redhat.com","date":"2019-07-01 06:50:49.000000000","tz":600},"committer":{"name":"Ian Wienand","email":"iwienand@redhat.com","date":"2019-07-01 07:49:46.000000000","tz":600},"subject":"Switch ovh/GRA1 to opendev.org mirror","message":"Switch ovh/GRA1 to opendev.org mirror\n\nThis is our second kafs based mirror (rax.IAD being the first), chosen\nto be geographically remote to the AFS servers.  It is currently\nmanually configured with 5.2-rc7 kernel, which has all the current\nkAFS fixes incorporated.\n\nSwitch it in for initial testing\n\nChange-Id: I83c73a8cff9728336b6208cab04e7c97c4c4ad86\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/45f3b85860123129630b612ef1f28c5456875892"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/45f3b85860123129630b612ef1f28c5456875892"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
