)]}'
{"id":"openstack%2Fproject-config~668681","triplet_id":"openstack%2Fproject-config~master~I0bd5ba5404989ca7ad4f62afe6ac177e1d95f45b","project":"openstack/project-config","branch":"master","topic":"add-microstack","hashtags":[],"change_id":"I0bd5ba5404989ca7ad4f62afe6ac177e1d95f45b","subject":"Add microstack project.","status":"MERGED","created":"2019-07-02 14:18:23.000000000","updated":"2019-07-17 17:52:45.000000000","submitted":"2019-07-17 17:52:45.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"668681-1563385965516-6f1d36b5","meta_rev_id":"f4e5ebc3a6ad3a7aeaac4c5137b7c2826ac1228c","_number":668681,"virtual_id_number":668681,"owner":{"_account_id":26040,"name":"Pete Vander Giessen","email":"pete.vandergiessen@canonical.com","username":"petevg"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":1,"name":"James E. Blair","email":"jim@acmegating.com","username":"corvus"},{"value":0,"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":0,"_account_id":935,"name":"James Page","email":"ringo.page@gmail.com","username":"james-page"},{"value":0,"_account_id":20635,"name":"Ryan Beisner","email":"ryanbeisner@gmail.com","username":"ryan-beisner"},{"value":2,"date":"2019-07-17 17:52:45.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":28543,"name":"shupeng","email":"15050873171@163.com","username":"Vieri"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":1,"name":"James E. Blair","email":"jim@acmegating.com","username":"corvus"},"all":[{"value":2,"date":"2019-07-17 17:40:08.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":1,"name":"James E. Blair","email":"jim@acmegating.com","username":"corvus"},{"value":2,"date":"2019-07-12 06:01:14.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":1,"date":"2019-07-11 16:32:28.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":935,"name":"James Page","email":"ringo.page@gmail.com","username":"james-page"},{"value":1,"date":"2019-07-11 17:00:21.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":20635,"name":"Ryan Beisner","email":"ryanbeisner@gmail.com","username":"ryan-beisner"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2019-07-12 07:08:41.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":28543,"name":"shupeng","email":"15050873171@163.com","username":"Vieri"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":1,"name":"James E. Blair","email":"jim@acmegating.com","username":"corvus"},"all":[{"value":1,"date":"2019-07-17 17:40:08.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":1,"name":"James E. Blair","email":"jim@acmegating.com","username":"corvus"},{"value":0,"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":0,"_account_id":935,"name":"James Page","email":"ringo.page@gmail.com","username":"james-page"},{"value":0,"_account_id":20635,"name":"Ryan Beisner","email":"ryanbeisner@gmail.com","username":"ryan-beisner"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":28543,"name":"shupeng","email":"15050873171@163.com","username":"Vieri"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1,"name":"James E. Blair","email":"jim@acmegating.com","username":"corvus"},{"_account_id":935,"name":"James Page","email":"ringo.page@gmail.com","username":"james-page"},{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"_account_id":20635,"name":"Ryan Beisner","email":"ryanbeisner@gmail.com","username":"ryan-beisner"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28543,"name":"shupeng","email":"15050873171@163.com","username":"Vieri"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-07-11 16:32:28.000000000","updated_by":{"_account_id":935,"name":"James Page","email":"ringo.page@gmail.com","username":"james-page"},"reviewer":{"_account_id":935,"name":"James Page","email":"ringo.page@gmail.com","username":"james-page"},"state":"REVIEWER"},{"updated":"2019-07-11 17:00:21.000000000","updated_by":{"_account_id":20635,"name":"Ryan Beisner","email":"ryanbeisner@gmail.com","username":"ryan-beisner"},"reviewer":{"_account_id":20635,"name":"Ryan Beisner","email":"ryanbeisner@gmail.com","username":"ryan-beisner"},"state":"REVIEWER"},{"updated":"2019-07-12 06:01:14.000000000","updated_by":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"reviewer":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"state":"REVIEWER"},{"updated":"2019-07-12 07:08:41.000000000","updated_by":{"_account_id":28543,"name":"shupeng","email":"15050873171@163.com","username":"Vieri"},"reviewer":{"_account_id":28543,"name":"shupeng","email":"15050873171@163.com","username":"Vieri"},"state":"REVIEWER"},{"updated":"2019-07-17 17:40:08.000000000","updated_by":{"_account_id":1,"name":"James E. Blair","email":"jim@acmegating.com","username":"corvus"},"reviewer":{"_account_id":1,"name":"James E. Blair","email":"jim@acmegating.com","username":"corvus"},"state":"REVIEWER"},{"updated":"2019-07-17 17:52:45.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"5b09f49c8e966708302eea2db2875a3364313aa0","author":{"_account_id":26040,"name":"Pete Vander Giessen","email":"pete.vandergiessen@canonical.com","username":"petevg"},"date":"2019-07-02 14:18:23.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"914f053c485af63869a2b0c27701ef4811a9138f","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-02 14:34:51.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-zuul-jobs-linters http://logs.openstack.org/81/668681/1/check/openstack-zuul-jobs-linters/c06b9b2/ : SUCCESS in 9m 13s\n- project-config-gerrit http://logs.openstack.org/81/668681/1/check/project-config-gerrit/545fbce/ : FAILURE in 2m 54s","accounts_in_message":[],"_revision_number":1},{"id":"cf0125f478ddb2d6fed771fa4b575316962e398e","author":{"_account_id":26040,"name":"Pete Vander Giessen","email":"pete.vandergiessen@canonical.com","username":"petevg"},"date":"2019-07-11 16:09:14.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"61be575570a8eab54f2414ba48df5841b313d3ae","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-11 16:22:58.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-zuul-jobs-linters http://logs.openstack.org/81/668681/2/check/openstack-zuul-jobs-linters/d3b342f/ : SUCCESS in 6m 09s\n- project-config-gerrit http://logs.openstack.org/81/668681/2/check/project-config-gerrit/cb4f2ca/ : SUCCESS in 4m 16s","accounts_in_message":[],"_revision_number":2},{"id":"1671b345e27c6bed857010cb3a45f7ec0c02629f","author":{"_account_id":935,"name":"James Page","email":"ringo.page@gmail.com","username":"james-page"},"date":"2019-07-11 16:32:28.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"268f7954f95b9010bf75bec5aa62725800827411","author":{"_account_id":20635,"name":"Ryan Beisner","email":"ryanbeisner@gmail.com","username":"ryan-beisner"},"date":"2019-07-11 17:00:21.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"c23b3b4bcaf1875c74e7b2ca1a69c4e2fa5c974e","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2019-07-12 06:01:14.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"7010d9a140a495f6afc282ca0b655428d05deace","author":{"_account_id":28543,"name":"shupeng","email":"15050873171@163.com","username":"Vieri"},"date":"2019-07-12 07:08:41.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"9ed0ef0a7b94feeda0960b7c1ee97d85ed89ad6b","author":{"_account_id":1,"name":"James E. Blair","email":"jim@acmegating.com","username":"corvus"},"date":"2019-07-17 17:40:08.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"a613e4b4dae84ec69fc21341a370cc09846d32b7","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-17 17:40:21.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"3bca7b336c30fc281a341397d6dd8f21144ef2f4","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-17 17:52:45.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"38d74b6c5d72bd4fc08d7d3498839ecd5dcf7101","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-17 17:52:45.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-zuul-jobs-linters http://logs.openstack.org/81/668681/2/gate/openstack-zuul-jobs-linters/6ba08c8/ : SUCCESS in 6m 44s\n- project-config-gerrit http://logs.openstack.org/81/668681/2/gate/project-config-gerrit/c33294a/ : SUCCESS in 4m 17s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"bc0d805e89b4151918d3c8ca1b93087e9a1c8407","revisions":{"ad3cf3fc224ac31f4f57f75c6ed39469ed9edee5":{"kind":"REWORK","_number":1,"created":"2019-07-02 14:18:23.000000000","uploader":{"_account_id":26040,"name":"Pete Vander Giessen","email":"pete.vandergiessen@canonical.com","username":"petevg"},"ref":"refs/changes/81/668681/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/81/668681/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/81/668681/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/81/668681/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/81/668681/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/81/668681/1"}}},"commit":{"parents":[{"commit":"59e9beeea67f086c5a52d86237fb4818cbec5d34","subject":"Merge \"Add testing job to zuul-jobs\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/59e9beeea67f086c5a52d86237fb4818cbec5d34"}]}],"author":{"name":"Pete Vander Giessen","email":"pete.vandergiessen@canonical.com","date":"2019-07-02 14:16:50.000000000","tz":-240},"committer":{"name":"Pete Vander Giessen","email":"pete.vandergiessen@canonical.com","date":"2019-07-02 14:16:50.000000000","tz":-240},"subject":"Add microstack project.","message":"Add microstack project.\n\nmicrostack is a basic Openstack deploy packaged in a snap!\n\nIt\u0027s currently hosted on github, but we\u0027re moving it to opendev to\ntake advantage of the zuul gate, among other things.\n\nChange-Id: I0bd5ba5404989ca7ad4f62afe6ac177e1d95f45b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/ad3cf3fc224ac31f4f57f75c6ed39469ed9edee5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/ad3cf3fc224ac31f4f57f75c6ed39469ed9edee5"}]},"branch":"refs/heads/master"},"bc0d805e89b4151918d3c8ca1b93087e9a1c8407":{"kind":"REWORK","_number":2,"created":"2019-07-11 16:09:14.000000000","uploader":{"_account_id":26040,"name":"Pete Vander Giessen","email":"pete.vandergiessen@canonical.com","username":"petevg"},"ref":"refs/changes/81/668681/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/81/668681/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/81/668681/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/81/668681/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/81/668681/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/81/668681/2"}}},"commit":{"parents":[{"commit":"59e9beeea67f086c5a52d86237fb4818cbec5d34","subject":"Merge \"Add testing job to zuul-jobs\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/59e9beeea67f086c5a52d86237fb4818cbec5d34"}]}],"author":{"name":"Pete Vander Giessen","email":"pete.vandergiessen@canonical.com","date":"2019-07-02 14:16:50.000000000","tz":-240},"committer":{"name":"Pete Vander Giessen","email":"pete.vandergiessen@canonical.com","date":"2019-07-11 16:09:03.000000000","tz":-240},"subject":"Add microstack project.","message":"Add microstack project.\n\nmicrostack is a basic Openstack deploy packaged in a snap!\n\nIt\u0027s currently hosted on github, but we\u0027re moving it to opendev to\ntake advantage of the zuul gate, among other things.\n\nChange-Id: I0bd5ba5404989ca7ad4f62afe6ac177e1d95f45b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/bc0d805e89b4151918d3c8ca1b93087e9a1c8407"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/bc0d805e89b4151918d3c8ca1b93087e9a1c8407"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
