)]}'
{"id":"openstack%2Fproject-config~668952","triplet_id":"openstack%2Fproject-config~master~Iab6cb4039aaaba2773fc1a92d341b707b8bd99ba","project":"openstack/project-config","branch":"master","topic":"fortnebula-images","hashtags":[],"change_id":"Iab6cb4039aaaba2773fc1a92d341b707b8bd99ba","subject":"Upload test images to fortnebula cloud","status":"MERGED","created":"2019-07-03 16:03:47.000000000","updated":"2019-07-03 16:27:34.000000000","submitted":"2019-07-03 16:27:34.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"668952-1562171254989-577aa814","meta_rev_id":"02c19cb084b6b9bf64b450bd40a17e80c9d19af5","_number":668952,"virtual_id_number":668952,"owner":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":1,"name":"James E. Blair","email":"jim@acmegating.com","username":"corvus"},{"value":0,"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},{"value":2,"date":"2019-07-03 16:27:34.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":1,"name":"James E. Blair","email":"jim@acmegating.com","username":"corvus"},"all":[{"value":2,"date":"2019-07-03 16:08:05.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":1,"name":"James E. Blair","email":"jim@acmegating.com","username":"corvus"},{"value":2,"date":"2019-07-03 16:09:39.000000000","_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},"all":[{"value":0,"_account_id":1,"name":"James E. Blair","email":"jim@acmegating.com","username":"corvus"},{"value":1,"date":"2019-07-03 16:09:39.000000000","_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1,"name":"James E. Blair","email":"jim@acmegating.com","username":"corvus"},{"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-07-03 16:08:05.000000000","updated_by":{"_account_id":1,"name":"James E. Blair","email":"jim@acmegating.com","username":"corvus"},"reviewer":{"_account_id":1,"name":"James E. Blair","email":"jim@acmegating.com","username":"corvus"},"state":"REVIEWER"},{"updated":"2019-07-03 16:09:39.000000000","updated_by":{"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},"reviewer":{"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},"state":"REVIEWER"},{"updated":"2019-07-03 16:27:34.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"5f16082273ba6c4d9e10c89a850b7c22b7f048e6","author":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"date":"2019-07-03 16:03:47.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"d641656dd7dd21516d3594d1df097bf60c943b01","author":{"_account_id":1,"name":"James E. Blair","email":"jim@acmegating.com","username":"corvus"},"date":"2019-07-03 16:08:05.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"1c29884612407e70481e65039e0c9f825b789676","author":{"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},"date":"2019-07-03 16:09:39.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"6049d4e5256c36078fae734383f01b7f28e6e259","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-03 16:17:29.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-zuul-jobs-linters http://logs.openstack.org/52/668952/1/check/openstack-zuul-jobs-linters/6fde876/ : SUCCESS in 7m 04s\n- project-config-nodepool http://logs.openstack.org/52/668952/1/check/project-config-nodepool/e6b7fd8/ : SUCCESS in 3m 31s","accounts_in_message":[],"_revision_number":1},{"id":"5db68cc4021c8cb6a8113d293d09ecfd556a04b7","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-03 16:17:54.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"bc63ef401d007b0fb8d8f8e922a6d470f38783ea","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-03 16:27:34.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"edaa9c21e5759ca5eb867212a1978b994ded07ce","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-03 16:27:34.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-zuul-jobs-linters http://logs.openstack.org/52/668952/1/gate/openstack-zuul-jobs-linters/2ddcfca/ : SUCCESS in 7m 18s\n- project-config-nodepool http://logs.openstack.org/52/668952/1/gate/project-config-nodepool/2a6daa8/ : SUCCESS in 3m 16s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"3ced74e1f71e1ed705fb8e6c1d7d7fa523aaa82a","revisions":{"3ced74e1f71e1ed705fb8e6c1d7d7fa523aaa82a":{"kind":"REWORK","_number":1,"created":"2019-07-03 16:03:47.000000000","uploader":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"ref":"refs/changes/52/668952/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/52/668952/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/52/668952/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/52/668952/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/52/668952/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/52/668952/1"}}},"commit":{"parents":[{"commit":"e42d998f9f8e5e79ba5c95f5b363e9efd9fff131","subject":"Merge \"Remove zuul-jobs config\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/e42d998f9f8e5e79ba5c95f5b363e9efd9fff131"}]}],"author":{"name":"Clark Boylan","email":"clark.boylan@gmail.com","date":"2019-07-03 16:03:15.000000000","tz":-420},"committer":{"name":"Clark Boylan","email":"clark.boylan@gmail.com","date":"2019-07-03 16:03:15.000000000","tz":-420},"subject":"Upload test images to fortnebula cloud","message":"Upload test images to fortnebula cloud\n\nThis should be safe to get running prior to the mirror being fully\ndeployed. We only need the mirror when we add test nodes.\n\nChange-Id: Iab6cb4039aaaba2773fc1a92d341b707b8bd99ba\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/3ced74e1f71e1ed705fb8e6c1d7d7fa523aaa82a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/3ced74e1f71e1ed705fb8e6c1d7d7fa523aaa82a"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
