)]}'
{"id":"openstack%2Fproject-config~669231","triplet_id":"openstack%2Fproject-config~master~Ia12096af302f61e437e2b65c900c34da1b2a9d59","project":"openstack/project-config","branch":"master","topic":"disable-kafs","hashtags":[],"change_id":"Ia12096af302f61e437e2b65c900c34da1b2a9d59","subject":"Disable kafs servers for a bit","status":"MERGED","created":"2019-07-05 01:54:03.000000000","updated":"2019-07-05 12:51:52.000000000","submitted":"2019-07-05 02:10:50.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"669231-1562292650010-b16e9fe7","meta_rev_id":"49d3f8167d336a5ea004043e3d87ab922d5c825f","_number":669231,"virtual_id_number":669231,"owner":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},{"value":2,"date":"2019-07-05 02:10:49.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},"all":[{"value":2,"date":"2019-07-05 01:55:39.000000000","_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"all":[{"value":0,"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2019-07-05 02:02:48.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-07-05 01:55:39.000000000","updated_by":{"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},"reviewer":{"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},"state":"REVIEWER"},{"updated":"2019-07-05 02:10:49.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"7fe5258a359b174271cffa799670c40fe6e64246","author":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"date":"2019-07-05 01:54:03.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"60ed546fefdad924d11041671a342265d6ef4dc1","author":{"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},"date":"2019-07-05 01:55:39.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"942d1b0c19472dd2f9b2cfaa90a759ebce08ec2e","author":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"date":"2019-07-05 02:02:48.000000000","message":"Patch Set 1: Workflow+1\n\ntaking the liberty of merging this, because the reason we\u0027re switching it out is not everyone is around :)","accounts_in_message":[],"_revision_number":1},{"id":"aa5cc83acee0c592ebcce0eaae731dcb70f291e3","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-05 02:03:01.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-zuul-jobs-linters http://logs.openstack.org/31/669231/1/check/openstack-zuul-jobs-linters/26583d5/ : SUCCESS in 7m 27s","accounts_in_message":[],"_revision_number":1},{"id":"ace51554cb647ff9981c23366b334fbda5e4ba0e","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-05 02:03:15.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"07ea9a3f1b9072c1ed904c80f8ce72dd7ebdf17b","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-05 02:10:49.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-zuul-jobs-linters http://logs.openstack.org/31/669231/1/gate/openstack-zuul-jobs-linters/2e31906/ : SUCCESS in 6m 12s","accounts_in_message":[],"_revision_number":1},{"id":"0435742a43cd3af4c68ce533f3c0ae88549d6ea2","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-05 02:10:50.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"da26961ccf37197b13211f687390212901a306e3","author":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"date":"2019-07-05 12:51:52.000000000","message":"Patch Set 1:\n\nHas my (post-hoc) +2, thanks!","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"e0388243d3a617fb72e92cd1d0f10aff43b60fcd","revisions":{"e0388243d3a617fb72e92cd1d0f10aff43b60fcd":{"kind":"REWORK","_number":1,"created":"2019-07-05 01:54:03.000000000","uploader":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"ref":"refs/changes/31/669231/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/31/669231/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/31/669231/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/31/669231/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/31/669231/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/31/669231/1"}}},"commit":{"parents":[{"commit":"f47578a938a8fac1d1206eef49597ec1b25d533f","subject":"Normalize projects.yaml","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/f47578a938a8fac1d1206eef49597ec1b25d533f"}]}],"author":{"name":"Ian Wienand","email":"iwienand@redhat.com","date":"2019-07-05 01:52:48.000000000","tz":600},"committer":{"name":"Ian Wienand","email":"iwienand@redhat.com","date":"2019-07-05 01:52:48.000000000","tz":600},"subject":"Disable kafs servers for a bit","message":"Disable kafs servers for a bit\n\nWe have some new changes to try that need to be rolled out.  In the\nmean time, they have a tendency to hard crash in fscahe which needs\nvery active monitoring which isn\u0027t happening right now.\n\nChange-Id: Ia12096af302f61e437e2b65c900c34da1b2a9d59\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/e0388243d3a617fb72e92cd1d0f10aff43b60fcd"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/e0388243d3a617fb72e92cd1d0f10aff43b60fcd"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
