)]}'
{"id":"openstack%2Fproject-config~669632","triplet_id":"openstack%2Fproject-config~master~Iae769fd82b5f94ddbcc40369df3317593c49bbbc","project":"openstack/project-config","branch":"master","topic":"clean-up-retired-charm-ceph","hashtags":[],"change_id":"Iae769fd82b5f94ddbcc40369df3317593c49bbbc","subject":"Add charm-ceph to allow cleaning up retired project","status":"ABANDONED","created":"2019-07-08 10:42:00.000000000","updated":"2019-07-08 10:55:54.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"d5a1d7c1f210f915bbb4154ae8a9c65bf1e8bc70","_number":669632,"virtual_id_number":669632,"owner":{"_account_id":20634,"name":"Chris MacNaughton","email":"chris.macnaughton@canonical.com","username":"Chris.MacNaughton"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":1,"date":"2019-07-08 10:54:22.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"disliked":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"all":[{"value":-1,"date":"2019-07-08 10:47:21.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":-1,"default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-07-08 10:47:21.000000000","updated_by":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"reviewer":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"state":"REVIEWER"},{"updated":"2019-07-08 10:54:22.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"9f3bd4f46a45a5330bc5a45603f6dea8b1479573","author":{"_account_id":20634,"name":"Chris MacNaughton","email":"chris.macnaughton@canonical.com","username":"Chris.MacNaughton"},"date":"2019-07-08 10:42:00.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"4a8af107e23289340e1783734e2b09643b9f034a","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2019-07-08 10:47:21.000000000","message":"Patch Set 1: Code-Review-1\n\nYou need in addition:\n* update ACLs, they should be read-only.\n* Add noop-jobs to zuul.d/projects.yaml","accounts_in_message":[],"_revision_number":1},{"id":"526b44a2418728753a1e7453fc1ff59d079ddc45","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-08 10:54:22.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-zuul-jobs-linters http://logs.openstack.org/32/669632/1/check/openstack-zuul-jobs-linters/96c1c7e/ : SUCCESS in 5m 59s\n- project-config-gerrit http://logs.openstack.org/32/669632/1/check/project-config-gerrit/c2209e0/ : SUCCESS in 3m 43s","accounts_in_message":[],"_revision_number":1},{"id":"faa28bbd313086d1781c6401e03326e29a524914","author":{"_account_id":20634,"name":"Chris MacNaughton","email":"chris.macnaughton@canonical.com","username":"Chris.MacNaughton"},"date":"2019-07-08 10:55:54.000000000","message":"Abandoned\n\nAbandoning per suggestion on the openstack-discuss mailing list (http://lists.openstack.org/pipermail/openstack-discuss/2019-July/007582.html)","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"8801f74defa657bd923a5026890e1ef4e4070220","revisions":{"8801f74defa657bd923a5026890e1ef4e4070220":{"kind":"REWORK","_number":1,"created":"2019-07-08 10:42:00.000000000","uploader":{"_account_id":20634,"name":"Chris MacNaughton","email":"chris.macnaughton@canonical.com","username":"Chris.MacNaughton"},"ref":"refs/changes/32/669632/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/32/669632/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/32/669632/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/32/669632/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/32/669632/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/32/669632/1"}}},"commit":{"parents":[{"commit":"69b6379e677a3e11dadbff1795c3796f015f4395","subject":"Move gerritbot to #openstack-kolla for kayobe","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/69b6379e677a3e11dadbff1795c3796f015f4395"}]}],"author":{"name":"Chris MacNaughton","email":"chris.macnaughton@canonical.com","date":"2019-07-08 10:41:51.000000000","tz":120},"committer":{"name":"Chris MacNaughton","email":"chris.macnaughton@canonical.com","date":"2019-07-08 10:41:51.000000000","tz":120},"subject":"Add charm-ceph to allow cleaning up retired project","message":"Add charm-ceph to allow cleaning up retired project\n\nChange-Id: Iae769fd82b5f94ddbcc40369df3317593c49bbbc\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/8801f74defa657bd923a5026890e1ef4e4070220"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/8801f74defa657bd923a5026890e1ef4e4070220"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
