)]}'
{"id":"openstack%2Fproject-config~669883","triplet_id":"openstack%2Fproject-config~master~Ib50a7a90c8049537663da7c9b68f7e68daef5a83","project":"openstack/project-config","branch":"master","topic":"retire-tempest-tripleo-ui","hashtags":[],"change_id":"Ib50a7a90c8049537663da7c9b68f7e68daef5a83","subject":"Drop tempest-tripleo-ui","status":"MERGED","created":"2019-07-09 14:44:06.000000000","updated":"2019-07-10 16:41:16.000000000","submitted":"2019-07-10 16:41:15.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"669883-1562776876283-59304768","meta_rev_id":"87aec012fd330982f0453c6664535a7bf54a5ea9","_number":669883,"virtual_id_number":669883,"owner":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"value":0,"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":2,"date":"2019-07-10 16:41:15.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"all":[{"value":2,"date":"2019-07-10 16:28:54.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"value":2,"date":"2019-07-09 17:25:27.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"all":[{"value":1,"date":"2019-07-10 16:28:54.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"value":0,"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-07-09 17:25:27.000000000","updated_by":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"reviewer":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"state":"REVIEWER"},{"updated":"2019-07-10 16:28:54.000000000","updated_by":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"reviewer":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"state":"REVIEWER"},{"updated":"2019-07-10 16:41:15.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"008e7812198d95cf2aa23fc9e915586e93065046","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2019-07-09 14:44:06.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"ab777a26daf7452133eb2658090c5b5c8966f096","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-09 15:02:17.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-zuul-jobs-linters http://logs.openstack.org/83/669883/1/check/openstack-zuul-jobs-linters/6bcabb9/ : SUCCESS in 7m 07s\n- project-config-gerrit http://logs.openstack.org/83/669883/1/check/project-config-gerrit/d4f1c7c/ : SUCCESS in 3m 41s","accounts_in_message":[],"_revision_number":1},{"id":"85ac08eca54d9c8b3aaf012594cb7abd196dc356","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2019-07-09 16:51:32.000000000","message":"Patch Set 1: Code-Review-1\n\ngerrit/projects.yaml is already updated but we missed something:\n\n- project: openstack/tempest-tripleo-ui\n  description: RETIRED TripleO UI plugin for tempest\n  acl-config: /home/gerrit2/acls/openstack/retired.config\n  groups:\n    - tripleo\n\nCould you remove the \"groups: -tripleo\" lines, please? Then this can go in...","accounts_in_message":[],"_revision_number":1},{"id":"2e6bf559703df372667328a79ebc225797356a3a","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2019-07-09 17:14:39.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"adb18c6183ac41d2a524e8ff8ebcac461db3e611","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2019-07-09 17:25:27.000000000","message":"Patch Set 2: Code-Review+2\n\nthanks","accounts_in_message":[],"_revision_number":2},{"id":"94b5f028c0b66e5c8d5c9777b7ef9b60606dafc8","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-09 17:27:56.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-zuul-jobs-linters http://logs.openstack.org/83/669883/2/check/openstack-zuul-jobs-linters/745dcca/ : SUCCESS in 6m 05s\n- project-config-gerrit http://logs.openstack.org/83/669883/2/check/project-config-gerrit/bc1fa03/ : SUCCESS in 3m 23s","accounts_in_message":[],"_revision_number":2},{"id":"7458d0ba10505a7d8fe03abfdd7dc170f026282f","author":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"date":"2019-07-10 16:28:54.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"5c6cf48576458fdce324317ef2c63760c730ecd4","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-10 16:29:13.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"76423ff8bfd4b3915ab56648f32d8689c9117800","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-10 16:41:15.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-zuul-jobs-linters http://logs.openstack.org/83/669883/2/gate/openstack-zuul-jobs-linters/908097b/ : SUCCESS in 6m 12s\n- project-config-gerrit http://logs.openstack.org/83/669883/2/gate/project-config-gerrit/5100d83/ : SUCCESS in 3m 51s","accounts_in_message":[],"_revision_number":2},{"id":"bec86042ca78c87105cda8b1cda6630d06c6ba9f","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-10 16:41:16.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"4b3b3b8d767e8297ba5ed8187740402113424c36","revisions":{"bc7d0789e1b62bff89773a6ee0686ab3bba82070":{"kind":"REWORK","_number":1,"created":"2019-07-09 14:44:06.000000000","uploader":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"ref":"refs/changes/83/669883/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/83/669883/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/83/669883/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/83/669883/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/83/669883/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/83/669883/1"}}},"commit":{"parents":[{"commit":"d6fb699719fd2eb379ec380f2223e6db68bb073e","subject":"Merge \"Mark networking-ovn-tempest-dsvm-ovs-release as voting\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/d6fb699719fd2eb379ec380f2223e6db68bb073e"}]}],"author":{"name":"Alex Schultz","email":"aschultz@redhat.com","date":"2019-07-09 14:43:30.000000000","tz":-360},"committer":{"name":"Alex Schultz","email":"aschultz@redhat.com","date":"2019-07-09 14:43:30.000000000","tz":-360},"subject":"Drop tempest-tripleo-ui","message":"Drop tempest-tripleo-ui\n\nThe code has been retired, need to cleanup tempest-tripleo-ui from\nprojects.yaml and main.yaml\n\nChange-Id: Ib50a7a90c8049537663da7c9b68f7e68daef5a83\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/bc7d0789e1b62bff89773a6ee0686ab3bba82070"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/bc7d0789e1b62bff89773a6ee0686ab3bba82070"}]},"branch":"refs/heads/master"},"4b3b3b8d767e8297ba5ed8187740402113424c36":{"kind":"REWORK","_number":2,"created":"2019-07-09 17:14:39.000000000","uploader":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"ref":"refs/changes/83/669883/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/83/669883/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/83/669883/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/83/669883/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/83/669883/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/83/669883/2"}}},"commit":{"parents":[{"commit":"d6fb699719fd2eb379ec380f2223e6db68bb073e","subject":"Merge \"Mark networking-ovn-tempest-dsvm-ovs-release as voting\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/d6fb699719fd2eb379ec380f2223e6db68bb073e"}]}],"author":{"name":"Alex Schultz","email":"aschultz@redhat.com","date":"2019-07-09 14:43:30.000000000","tz":-360},"committer":{"name":"Alex Schultz","email":"aschultz@redhat.com","date":"2019-07-09 17:14:36.000000000","tz":-360},"subject":"Drop tempest-tripleo-ui","message":"Drop tempest-tripleo-ui\n\nThe code has been retired, need to cleanup tempest-tripleo-ui from\nprojects.yaml and main.yaml\n\nChange-Id: Ib50a7a90c8049537663da7c9b68f7e68daef5a83\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/4b3b3b8d767e8297ba5ed8187740402113424c36"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/4b3b3b8d767e8297ba5ed8187740402113424c36"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
