)]}'
{"id":"openstack%2Fproject-config~670269","triplet_id":"openstack%2Fproject-config~master~Idefab0b104cf05f17fcf5c756f5e56f0ecc3e79e","project":"openstack/project-config","branch":"master","topic":"retire-docs","hashtags":[],"change_id":"Idefab0b104cf05f17fcf5c756f5e56f0ecc3e79e","subject":"Update ACLs for moved doc projects","status":"MERGED","created":"2019-07-11 10:27:42.000000000","updated":"2019-10-28 18:04:37.000000000","submitted":"2019-10-28 18:04:37.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":7,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"670269-1572285877152-cd6aae0a","meta_rev_id":"d1559c1fce176ff62c0aca7198dcdf6327ffc630","_number":670269,"virtual_id_number":670269,"owner":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},{"value":0,"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":0,"_account_id":8099,"name":"Graham Hayes","email":"gr@ham.ie","username":"graham"},{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2019-10-28 18:04:36.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"recommended":{"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},"all":[{"value":1,"date":"2019-10-28 17:30:30.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},{"value":2,"date":"2019-10-28 17:28:22.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":0,"_account_id":8099,"name":"Graham Hayes","email":"gr@ham.ie","username":"graham"},{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2019-10-28 17:55:25.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"all":[{"value":0,"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},{"value":0,"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":0,"_account_id":8099,"name":"Graham Hayes","email":"gr@ham.ie","username":"graham"},{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2019-10-28 17:55:25.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},{"_account_id":8099,"name":"Graham Hayes","email":"gr@ham.ie","username":"graham"},{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-07-11 10:29:19.000000000","updated_by":{"_account_id":8099,"name":"Graham Hayes","email":"gr@ham.ie","username":"graham"},"reviewer":{"_account_id":8099,"name":"Graham Hayes","email":"gr@ham.ie","username":"graham"},"state":"REVIEWER"},{"updated":"2019-10-28 17:28:22.000000000","updated_by":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"reviewer":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"state":"REVIEWER"},{"updated":"2019-10-28 17:30:30.000000000","updated_by":{"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},"reviewer":{"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},"state":"REVIEWER"},{"updated":"2019-10-28 17:55:25.000000000","updated_by":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"reviewer":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"state":"REVIEWER"},{"updated":"2019-10-28 18:04:36.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"ad9d57621dd0c455c9ce24636ac415ffbd0dcab1","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2019-07-11 10:27:42.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"f0ec930a55d5282a7c81f97edce6b903854a68a7","author":{"_account_id":8099,"name":"Graham Hayes","email":"gr@ham.ie","username":"graham"},"date":"2019-07-11 10:29:19.000000000","message":"Patch Set 1: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"52ebed2f5e01a3649bfd96f7cf639150f524dc11","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2019-07-11 10:31:31.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"a1b1b4edb58979dab188e7a20ddfb4a1ddf77ffa","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-11 10:39:35.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-zuul-jobs-linters http://logs.openstack.org/69/670269/1/check/openstack-zuul-jobs-linters/7e7867c/ : SUCCESS in 6m 15s\n- project-config-gerrit http://logs.openstack.org/69/670269/1/check/project-config-gerrit/3cb445a/ : SUCCESS in 3m 37s","accounts_in_message":[],"_revision_number":1},{"id":"e9f91d9bda9c7051fa6ba21625e09548ad7bee8f","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2019-07-11 10:39:58.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"00cb96ba7cff79e9bb97dcee7e7d0399e11862a4","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-11 10:51:35.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-zuul-jobs-linters http://logs.openstack.org/69/670269/2/check/openstack-zuul-jobs-linters/1da78be/ : SUCCESS in 8m 29s\n- project-config-gerrit http://logs.openstack.org/69/670269/2/check/project-config-gerrit/1083082/ : SUCCESS in 4m 46s","accounts_in_message":[],"_revision_number":2},{"id":"2d0a239617d285f07822b3b0cce6943d3a46aae2","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2019-07-11 11:35:49.000000000","message":"Patch Set 2: Code-Review-1\n\n(1 comment)\n\nLet\u0027s not duplicate ACL files\n\nBtw. [1] in commit message has no link","accounts_in_message":[],"_revision_number":2},{"id":"434015409ddd125db22c34baf1611919083d123b","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2019-07-12 09:58:12.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"bad2c61ee545f7bce070bf351da7f8f0f1b30150","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-12 10:11:39.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-zuul-jobs-linters http://logs.openstack.org/69/670269/3/check/openstack-zuul-jobs-linters/f2af00d/ : SUCCESS in 7m 12s\n- project-config-gerrit http://logs.openstack.org/69/670269/3/check/project-config-gerrit/b8d64e5/ : FAILURE in 2m 48s","accounts_in_message":[],"_revision_number":3},{"id":"427592bc1f3950cb177ba4e33a9fa827dd6878ec","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2019-07-12 10:38:17.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"9a834a03d10b1e6c8a567ce34c64bddf8427caaf","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2019-10-26 19:04:07.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"b7c5a4b939689bfe6d4ca12cd6eb1ecfa628e3be","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-26 19:16:46.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/de5ae9041f444793b8eccc88125578e9 : SUCCESS in 6m 27s\n- project-config-gerrit https://zuul.opendev.org/t/openstack/build/3717678d294d42f98f34a74455a4d62d : SUCCESS in 3m 16s","accounts_in_message":[],"_revision_number":4},{"id":"d35cd9af4179eff4bfb66a7b5ef3ef8a1517266e","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2019-10-26 19:27:57.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"3b62e6bfff1df65fca4fe1ddca63128f8089f0d8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-26 19:33:56.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/596441774661482a8a29edd06db50c7d : SUCCESS in 5m 09s\n- project-config-gerrit https://zuul.opendev.org/t/openstack/build/72534e70aafb418aa76421f266be69bd : SUCCESS in 3m 27s","accounts_in_message":[],"_revision_number":5},{"id":"62241573ee4ad5b12d06bd0f25aebcce27279473","author":{"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},"date":"2019-10-28 17:18:02.000000000","message":"Patch Set 5:\n\n(1 comment)\n\nLooks fine, one question inline.","accounts_in_message":[],"_revision_number":5},{"id":"3c98de711ea49b2282a3f481f499976b69d209d8","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2019-10-28 17:20:14.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"a65fbcba5f354066ded7de9e78f25625d61f6f3f","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2019-10-28 17:28:22.000000000","message":"Patch Set 5: Code-Review+2","accounts_in_message":[],"_revision_number":5},{"id":"0073998ba0bc4ccbe3b30b8d54daba8880473389","author":{"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},"date":"2019-10-28 17:30:30.000000000","message":"Patch Set 5: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"86f35bf3e3810c32f6f9c7f346cf50d25d73ca27","author":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"date":"2019-10-28 17:55:25.000000000","message":"Patch Set 5: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":5},{"id":"d5b344f7cf9cbd6574a3f3df4ba34f2e489d9a10","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-28 17:55:37.000000000","message":"Patch Set 5: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":5},{"id":"f5387c1f7bc91e51f0902f49f64cd074e0ca54d1","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-28 18:04:36.000000000","message":"Patch Set 5: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/4a53fad0c05445428ee6519551878f32 : SUCCESS in 7m 32s\n- project-config-gerrit https://zuul.opendev.org/t/openstack/build/8dbda0a8dd5a46f49538692cc94a5d9d : SUCCESS in 4m 25s","accounts_in_message":[],"_revision_number":5},{"id":"eaecc77c8718adc5e0d480955689f9efc5bdcb14","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-28 18:04:37.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":5}],"current_revision_number":5,"current_revision":"bc78b6d26af4fba3c48d40b302778c4968de0fa7","revisions":{"90d6d364dfeca6a42f4090edf4f02e1cac27ea90":{"kind":"REWORK","_number":1,"created":"2019-07-11 10:27:42.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/69/670269/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/69/670269/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/69/670269/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/69/670269/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/69/670269/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/69/670269/1"}}},"commit":{"parents":[{"commit":"95b96765275f594cbd90abd0a4491bda4b7d3f23","subject":"Merge \"Retire docs-specs\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/95b96765275f594cbd90abd0a4491bda4b7d3f23"}]}],"author":{"name":"Stephen Finucane","email":"sfinucan@redhat.com","date":"2019-07-11 09:54:39.000000000","tz":60},"committer":{"name":"Stephen Finucane","email":"sfinucan@redhat.com","date":"2019-07-11 09:54:39.000000000","tz":60},"subject":"Update ACLs for moved doc projects","message":"Update ACLs for moved doc projects\n\nThis process was initiated in [1]. Make the Gerrit side of things\nhappen.\n\nChange-Id: Idefab0b104cf05f17fcf5c756f5e56f0ecc3e79e\nSigned-off-by: Stephen Finucane \u003csfinucan@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/90d6d364dfeca6a42f4090edf4f02e1cac27ea90"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/90d6d364dfeca6a42f4090edf4f02e1cac27ea90"}]},"branch":"refs/heads/master"},"5a37e43f92715a85c0fba55d0c54e40d65aafc9d":{"kind":"REWORK","_number":2,"created":"2019-07-11 10:39:58.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/69/670269/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/69/670269/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/69/670269/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/69/670269/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/69/670269/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/69/670269/2"}}},"commit":{"parents":[{"commit":"95b96765275f594cbd90abd0a4491bda4b7d3f23","subject":"Merge \"Retire docs-specs\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/95b96765275f594cbd90abd0a4491bda4b7d3f23"}]}],"author":{"name":"Stephen Finucane","email":"sfinucan@redhat.com","date":"2019-07-11 09:54:39.000000000","tz":60},"committer":{"name":"Stephen Finucane","email":"sfinucan@redhat.com","date":"2019-07-11 10:32:53.000000000","tz":60},"subject":"Update ACLs for moved doc projects","message":"Update ACLs for moved doc projects\n\nThis process was initiated in [1]. Make the Gerrit side of things\nhappen.\n\nChange-Id: Idefab0b104cf05f17fcf5c756f5e56f0ecc3e79e\nSigned-off-by: Stephen Finucane \u003csfinucan@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/5a37e43f92715a85c0fba55d0c54e40d65aafc9d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/5a37e43f92715a85c0fba55d0c54e40d65aafc9d"}]},"branch":"refs/heads/master"},"7d4a2d8f0ff3ae83d1d2ad9f21bbcee221ecc233":{"kind":"REWORK","_number":3,"created":"2019-07-12 09:58:12.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/69/670269/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/69/670269/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/69/670269/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/69/670269/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/69/670269/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/69/670269/3"}}},"commit":{"parents":[{"commit":"32a3f94ef3e12875377dc16485aef9295794caa4","subject":"Add shared \u0027oslo\u0027, \u0027oslo-independent\u0027 ACL files","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/32a3f94ef3e12875377dc16485aef9295794caa4"}]}],"author":{"name":"Stephen Finucane","email":"sfinucan@redhat.com","date":"2019-07-11 09:54:39.000000000","tz":60},"committer":{"name":"Stephen Finucane","email":"sfinucan@redhat.com","date":"2019-07-12 09:55:22.000000000","tz":60},"subject":"Update ACLs for moved doc projects","message":"Update ACLs for moved doc projects\n\nThis process was initiated in [1]. Make the Gerrit side of things\nhappen.\n\n[1] https://review.opendev.org/#/c/657141/\n\nChange-Id: Idefab0b104cf05f17fcf5c756f5e56f0ecc3e79e\nSigned-off-by: Stephen Finucane \u003csfinucan@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/7d4a2d8f0ff3ae83d1d2ad9f21bbcee221ecc233"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/7d4a2d8f0ff3ae83d1d2ad9f21bbcee221ecc233"}]},"branch":"refs/heads/master"},"36deebab440af73263f6cacd3bb12269968f4a0d":{"kind":"REWORK","_number":4,"created":"2019-10-26 19:04:07.000000000","uploader":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"ref":"refs/changes/69/670269/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/69/670269/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/69/670269/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/69/670269/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/69/670269/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/69/670269/4"}}},"commit":{"parents":[{"commit":"a2ee99a28c0e2004b562719711bffbc95dff2a36","subject":"Add shared \u0027oslo\u0027, \u0027oslo-independent\u0027 ACL files","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/a2ee99a28c0e2004b562719711bffbc95dff2a36"}]}],"author":{"name":"Stephen Finucane","email":"sfinucan@redhat.com","date":"2019-07-11 09:54:39.000000000","tz":60},"committer":{"name":"Andreas Jaeger","email":"aj@suse.com","date":"2019-10-26 19:01:02.000000000","tz":120},"subject":"Update ACLs for moved doc projects","message":"Update ACLs for moved doc projects\n\nThis process was initiated in [1]. Make the Gerrit side of things\nhappen.\n\nUpdate group for whereto, it should be part of openstack-doc-tools, not\nof documentation (which is part of the docs sig).\n\n[1] https://review.opendev.org/#/c/657141/\n\nChange-Id: Idefab0b104cf05f17fcf5c756f5e56f0ecc3e79e\nSigned-off-by: Stephen Finucane \u003csfinucan@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/36deebab440af73263f6cacd3bb12269968f4a0d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/36deebab440af73263f6cacd3bb12269968f4a0d"}]},"branch":"refs/heads/master"},"bc78b6d26af4fba3c48d40b302778c4968de0fa7":{"kind":"REWORK","_number":5,"created":"2019-10-26 19:27:57.000000000","uploader":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"ref":"refs/changes/69/670269/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/69/670269/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/69/670269/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/69/670269/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/69/670269/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/69/670269/5"}}},"commit":{"parents":[{"commit":"a2ee99a28c0e2004b562719711bffbc95dff2a36","subject":"Add shared \u0027oslo\u0027, \u0027oslo-independent\u0027 ACL files","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/a2ee99a28c0e2004b562719711bffbc95dff2a36"}]}],"author":{"name":"Stephen Finucane","email":"sfinucan@redhat.com","date":"2019-07-11 09:54:39.000000000","tz":60},"committer":{"name":"Andreas Jaeger","email":"aj@suse.com","date":"2019-10-26 19:27:03.000000000","tz":120},"subject":"Update ACLs for moved doc projects","message":"Update ACLs for moved doc projects\n\nThis process was initiated in [1]. Make the Gerrit side of things\nhappen.\n\nUpdate group for whereto, it should be part of openstack-doc-tools (on\nLaunchpad), not of documentation in Storyboard. Oslo does not use\nstoryboard currently and documentation is meant for Docs SIG.\n\n[1] https://review.opendev.org/#/c/657141/\n\nChange-Id: Idefab0b104cf05f17fcf5c756f5e56f0ecc3e79e\nSigned-off-by: Stephen Finucane \u003csfinucan@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/bc78b6d26af4fba3c48d40b302778c4968de0fa7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/bc78b6d26af4fba3c48d40b302778c4968de0fa7"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
