)]}'
{"id":"openstack%2Fproject-config~670369","triplet_id":"openstack%2Fproject-config~master~Ic3ea053093484fe6796f41a2d436bab6f79c8929","project":"openstack/project-config","branch":"master","topic":"zuulectomy","hashtags":[],"change_id":"Ic3ea053093484fe6796f41a2d436bab6f79c8929","subject":"Temporarily shadow base-jobs in openstack-zuul-jobs","status":"MERGED","created":"2019-07-11 18:58:28.000000000","updated":"2019-07-11 19:22:19.000000000","submitted":"2019-07-11 19:22:19.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"670369-1562872939691-77280bbb","meta_rev_id":"04982407a47352ca8c3c2e62ecf19264cbe08eae","_number":670369,"virtual_id_number":670369,"owner":{"_account_id":1,"name":"James E. Blair","email":"jim@acmegating.com","username":"corvus"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":2,"date":"2019-07-11 19:22:19.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"all":[{"value":2,"date":"2019-07-11 19:00:27.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2019-07-11 19:04:51.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"all":[{"value":0,"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2019-07-11 19:04:51.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-07-11 19:00:27.000000000","updated_by":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"reviewer":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"state":"REVIEWER"},{"updated":"2019-07-11 19:04:51.000000000","updated_by":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"reviewer":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"state":"REVIEWER"},{"updated":"2019-07-11 19:22:19.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"f3c64569822e1f60e8e200c5777cd70214599ba0","author":{"_account_id":1,"name":"James E. Blair","email":"jim@acmegating.com","username":"corvus"},"date":"2019-07-11 18:58:28.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"16c86fa93b19b4983eccfa347226906aaabf1494","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2019-07-11 19:00:27.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"3e0b9d2b824e2e4946cdd00f2ad95ec2ba391178","author":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"date":"2019-07-11 19:04:51.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"87dfae8ebada9149cc9d799fa241f55759551ea8","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-11 19:10:25.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-zuul-jobs-linters http://logs.openstack.org/69/670369/1/check/openstack-zuul-jobs-linters/f4e4d53/ : SUCCESS in 5m 59s\n- project-config-gerrit http://logs.openstack.org/69/670369/1/check/project-config-gerrit/5a1afc1/ : SUCCESS in 5m 06s","accounts_in_message":[],"_revision_number":1},{"id":"3ab9a9a9c9055a4a59f6f1b72650ee24c41d626e","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-11 19:10:40.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"78aeaa9aade49af598ff80de08b24bbeb8375e28","author":{"_account_id":1,"name":"James E. Blair","email":"jim@acmegating.com","username":"corvus"},"date":"2019-07-11 19:11:14.000000000","message":"Topic set to zuulectomy","accounts_in_message":[],"_revision_number":1},{"id":"93c4ce1fd6fb49c904a25161a7c2c29209483d3c","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-11 19:22:19.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-zuul-jobs-linters http://logs.openstack.org/69/670369/1/gate/openstack-zuul-jobs-linters/d13c017/ : SUCCESS in 6m 16s\n- project-config-gerrit http://logs.openstack.org/69/670369/1/gate/project-config-gerrit/2a8c22c/ : SUCCESS in 3m 42s","accounts_in_message":[],"_revision_number":1},{"id":"04982407a47352ca8c3c2e62ecf19264cbe08eae","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-11 19:22:19.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"6c2cbe510ff071cec977add5171646efdb966bcf","revisions":{"6c2cbe510ff071cec977add5171646efdb966bcf":{"kind":"REWORK","_number":1,"created":"2019-07-11 18:58:28.000000000","uploader":{"_account_id":1,"name":"James E. Blair","email":"jim@acmegating.com","username":"corvus"},"ref":"refs/changes/69/670369/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/69/670369/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/69/670369/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/69/670369/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/69/670369/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/69/670369/1"}}},"commit":{"parents":[{"commit":"816077b32e0f38877898b8b49709c1164eb13e1a","subject":"Fix typo in Zuul tenant config","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/816077b32e0f38877898b8b49709c1164eb13e1a"}]}],"author":{"name":"James E. Blair","email":"jeblair@redhat.com","date":"2019-07-11 18:57:01.000000000","tz":-420},"committer":{"name":"James E. Blair","email":"jeblair@redhat.com","date":"2019-07-11 18:58:18.000000000","tz":-420},"subject":"Temporarily shadow base-jobs in openstack-zuul-jobs","message":"Temporarily shadow base-jobs in openstack-zuul-jobs\n\nSo that we can move the nodeset definitions to opende/base-jobs,\nhave openstack-zuul-jobs shadow that repo.\n\nChange-Id: Ic3ea053093484fe6796f41a2d436bab6f79c8929\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/6c2cbe510ff071cec977add5171646efdb966bcf"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/6c2cbe510ff071cec977add5171646efdb966bcf"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
