)]}'
{"id":"openstack%2Fproject-config~670384","triplet_id":"openstack%2Fproject-config~master~I7664a5b9e104c3becdc04344d46308125cc2df1c","project":"openstack/project-config","branch":"master","topic":"zuulectomy","hashtags":[],"change_id":"I7664a5b9e104c3becdc04344d46308125cc2df1c","subject":"Add temporary shadow to opendev/base-jobs","status":"MERGED","created":"2019-07-11 20:06:25.000000000","updated":"2019-07-11 20:46:42.000000000","submitted":"2019-07-11 20:46:42.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"670384-1562878002154-ef2d6f53","meta_rev_id":"b03482f7bf734b8f5313ef83ba526affaae444fb","_number":670384,"virtual_id_number":670384,"owner":{"_account_id":1,"name":"James E. Blair","email":"jim@acmegating.com","username":"corvus"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":1,"name":"James E. Blair","email":"jim@acmegating.com","username":"corvus"},{"value":0,"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":2,"date":"2019-07-11 20:46:41.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":1,"name":"James E. Blair","email":"jim@acmegating.com","username":"corvus"},"all":[{"value":2,"date":"2019-07-11 20:34:58.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":1,"name":"James E. Blair","email":"jim@acmegating.com","username":"corvus"},{"value":2,"date":"2019-07-11 20:12:02.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":1,"name":"James E. Blair","email":"jim@acmegating.com","username":"corvus"},"all":[{"value":1,"date":"2019-07-11 20:34:58.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":1,"name":"James E. Blair","email":"jim@acmegating.com","username":"corvus"},{"value":0,"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1,"name":"James E. Blair","email":"jim@acmegating.com","username":"corvus"},{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-07-11 20:12:02.000000000","updated_by":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"reviewer":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"state":"REVIEWER"},{"updated":"2019-07-11 20:46:41.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"438d6bc12c764f3fcc81827aaf4138c64bb7d132","author":{"_account_id":1,"name":"James E. Blair","email":"jim@acmegating.com","username":"corvus"},"date":"2019-07-11 20:06:25.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"c4d6f5514bb592c7c3e575cb09af71891b6a1d46","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2019-07-11 20:12:02.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"d7fd1ea01be2090f4313d9b85087520542624bf0","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-11 20:23:29.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-zuul-jobs-linters http://logs.openstack.org/84/670384/1/check/openstack-zuul-jobs-linters/d6e23ea/ : SUCCESS in 6m 39s\n- project-config-gerrit http://logs.openstack.org/84/670384/1/check/project-config-gerrit/cef235d/ : SUCCESS in 3m 40s","accounts_in_message":[],"_revision_number":1},{"id":"e4cbb6879aef961f4de3e7296f278745ad7f02b2","author":{"_account_id":1,"name":"James E. Blair","email":"jim@acmegating.com","username":"corvus"},"date":"2019-07-11 20:34:58.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"312b0291188a1f99adcfb961688f665c573003b7","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-11 20:35:11.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"37e682bebb425964c75684711d674a837f304e10","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-11 20:46:41.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-zuul-jobs-linters http://logs.openstack.org/84/670384/1/gate/openstack-zuul-jobs-linters/65bf182/ : SUCCESS in 6m 20s\n- project-config-gerrit http://logs.openstack.org/84/670384/1/gate/project-config-gerrit/275605d/ : SUCCESS in 3m 41s","accounts_in_message":[],"_revision_number":1},{"id":"a26b3369b9ed4354f92d49c4f5b2822b9a480f02","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-11 20:46:42.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"bb6be03af076dde76eea87b3cf6a14cde02673f1","revisions":{"bb6be03af076dde76eea87b3cf6a14cde02673f1":{"kind":"REWORK","_number":1,"created":"2019-07-11 20:06:25.000000000","uploader":{"_account_id":1,"name":"James E. Blair","email":"jim@acmegating.com","username":"corvus"},"ref":"refs/changes/84/670384/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/84/670384/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/84/670384/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/84/670384/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/84/670384/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/84/670384/1"}}},"commit":{"parents":[{"commit":"b245c3003f62df0df010061f19590f2df1fe3b6e","subject":"Add some github projects to the Zuul tenant","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/b245c3003f62df0df010061f19590f2df1fe3b6e"}]}],"author":{"name":"James E. Blair","email":"jeblair@redhat.com","date":"2019-07-11 20:04:31.000000000","tz":-420},"committer":{"name":"James E. Blair","email":"jeblair@redhat.com","date":"2019-07-11 20:04:31.000000000","tz":-420},"subject":"Add temporary shadow to opendev/base-jobs","message":"Add temporary shadow to opendev/base-jobs\n\nAllow base-jobs to shadow project-config in the opendev tenant so\nthat we can move these nodeset definitions to base-jobs.\n\nChange-Id: I7664a5b9e104c3becdc04344d46308125cc2df1c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/bb6be03af076dde76eea87b3cf6a14cde02673f1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/bb6be03af076dde76eea87b3cf6a14cde02673f1"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
