)]}'
{"id":"openstack%2Fproject-config~670609","triplet_id":"openstack%2Fproject-config~master~I08b9663edb6938590400c4bbb77b27b538de5ffb","project":"openstack/project-config","branch":"master","hashtags":[],"change_id":"I08b9663edb6938590400c4bbb77b27b538de5ffb","subject":"Moving the workload on FNCI to 60%","status":"MERGED","created":"2019-07-12 17:38:59.000000000","updated":"2019-07-12 18:15:41.000000000","submitted":"2019-07-12 18:15:41.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"670609-1562955341283-77280bbb","meta_rev_id":"234f451582a80ab6ea9a4bcb27257fce79cc4abd","_number":670609,"virtual_id_number":670609,"owner":{"_account_id":20754,"name":"Donny Davis","email":"donny@fortnebula.com","username":"donnydavis"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"value":0,"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},{"value":2,"date":"2019-07-12 18:15:41.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"all":[{"value":2,"date":"2019-07-12 17:39:16.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"value":2,"date":"2019-07-12 17:42:35.000000000","_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},"all":[{"value":0,"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"value":1,"date":"2019-07-12 17:42:35.000000000","_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-07-12 17:39:16.000000000","updated_by":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"reviewer":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"state":"REVIEWER"},{"updated":"2019-07-12 17:42:35.000000000","updated_by":{"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},"reviewer":{"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},"state":"REVIEWER"},{"updated":"2019-07-12 18:15:41.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"1df6bb46dfae3c4d812c13c38d2cac77c5c4023c","author":{"_account_id":20754,"name":"Donny Davis","email":"donny@fortnebula.com","username":"donnydavis"},"date":"2019-07-12 17:38:59.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"6861735d333aa59a1458cf3c1e860f31ddf93223","author":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"date":"2019-07-12 17:39:16.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"6171fec97195351b606c7f95eca484b2059afaf6","author":{"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},"date":"2019-07-12 17:42:35.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"7048eba5d1125e745fe6d62241a90300ae4c9577","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-12 17:49:39.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-zuul-jobs-linters http://logs.openstack.org/09/670609/1/check/openstack-zuul-jobs-linters/b713cb1/ : SUCCESS in 6m 29s\n- project-config-nodepool http://logs.openstack.org/09/670609/1/check/project-config-nodepool/eb78580/ : SUCCESS in 4m 34s","accounts_in_message":[],"_revision_number":1},{"id":"781322a850680b5a4fa9812185fb4c58666d92e8","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-12 17:49:52.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"90207454be45f786888944a6d34d123d99400ed7","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-12 18:15:41.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-zuul-jobs-linters http://logs.openstack.org/09/670609/1/gate/openstack-zuul-jobs-linters/bf4db09/ : SUCCESS in 6m 09s\n- project-config-nodepool http://logs.openstack.org/09/670609/1/gate/project-config-nodepool/83d5c91/ : SUCCESS in 3m 33s","accounts_in_message":[],"_revision_number":1},{"id":"234f451582a80ab6ea9a4bcb27257fce79cc4abd","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-12 18:15:41.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"f6cddc8868b95a50902194d5e58bd9dc9a370e53","revisions":{"f6cddc8868b95a50902194d5e58bd9dc9a370e53":{"kind":"REWORK","_number":1,"created":"2019-07-12 17:38:59.000000000","uploader":{"_account_id":20754,"name":"Donny Davis","email":"donny@fortnebula.com","username":"donnydavis"},"ref":"refs/changes/09/670609/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/09/670609/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/09/670609/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/09/670609/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/09/670609/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/09/670609/1"}}},"commit":{"parents":[{"commit":"beab7eb21389810ec23549afbf18200da1811614","subject":"Scaling FNCI to 40 instances","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/beab7eb21389810ec23549afbf18200da1811614"}]}],"author":{"name":"Donny Davis","email":"donny@fortnebula.com","date":"2019-07-12 17:38:29.000000000","tz":-240},"committer":{"name":"Donny Davis","email":"donny@fortnebula.com","date":"2019-07-12 17:38:29.000000000","tz":-240},"subject":"Moving the workload on FNCI to 60%","message":"Moving the workload on FNCI to 60%\n\nChange-Id: I08b9663edb6938590400c4bbb77b27b538de5ffb\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/f6cddc8868b95a50902194d5e58bd9dc9a370e53"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/f6cddc8868b95a50902194d5e58bd9dc9a370e53"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
