)]}'
{"id":"openstack%2Fproject-config~670630","triplet_id":"openstack%2Fproject-config~master~I00c9f644eab73ef8b4189de4ea00891dca2c8f8c","project":"openstack/project-config","branch":"master","topic":"enable-limestone","hashtags":[],"change_id":"I00c9f644eab73ef8b4189de4ea00891dca2c8f8c","subject":"Reenable limestone cloud region in nodepool","status":"MERGED","created":"2019-07-12 20:01:29.000000000","updated":"2019-07-12 20:20:59.000000000","submitted":"2019-07-12 20:20:59.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"670630-1562962859091-77280bbb","meta_rev_id":"a65a19d8a8d4c70e8e6281f1fe7fb03be161766b","_number":670630,"virtual_id_number":670630,"owner":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":2,"date":"2019-07-12 20:20:58.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2019-07-12 20:03:09.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2019-07-12 20:03:09.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-07-12 20:03:09.000000000","updated_by":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"reviewer":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"state":"REVIEWER"},{"updated":"2019-07-12 20:20:58.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"1ab933de52403608360e8f794e95e04f67899f49","author":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"date":"2019-07-12 20:01:29.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"d87181aa6bb051dd873d876219cf8d27c7ea81b7","author":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"date":"2019-07-12 20:03:09.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\nSingle-core approving. I\u0027m around all evening and can revert if things go sideways again.","accounts_in_message":[],"_revision_number":1},{"id":"128944d5c6b8264109625d88036ea59dcbaf3b27","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-12 20:11:21.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-zuul-jobs-linters http://logs.openstack.org/30/670630/1/check/openstack-zuul-jobs-linters/8aeecc5/ : SUCCESS in 8m 17s\n- project-config-nodepool http://logs.openstack.org/30/670630/1/check/project-config-nodepool/11573eb/ : SUCCESS in 4m 46s","accounts_in_message":[],"_revision_number":1},{"id":"0d3d07a96abb9549618ba161225b9df43ce875ad","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-12 20:11:36.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"49ca021c05a057f143c22316ffe6a69e3d67b9de","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-12 20:20:58.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-zuul-jobs-linters http://logs.openstack.org/30/670630/1/gate/openstack-zuul-jobs-linters/dcbbffd/ : SUCCESS in 7m 40s\n- project-config-nodepool http://logs.openstack.org/30/670630/1/gate/project-config-nodepool/5ba961d/ : SUCCESS in 5m 17s","accounts_in_message":[],"_revision_number":1},{"id":"db647d6e1f513857c1f212c1d779df78a855cb20","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-12 20:20:59.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"553e143b728f493dc03b30d330f7cce9a04eaf13","revisions":{"553e143b728f493dc03b30d330f7cce9a04eaf13":{"kind":"REWORK","_number":1,"created":"2019-07-12 20:01:29.000000000","uploader":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"ref":"refs/changes/30/670630/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/30/670630/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/30/670630/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/30/670630/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/30/670630/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/30/670630/1"}}},"commit":{"parents":[{"commit":"fb8569a8d07810804bf51809386dce6ebab9e0ca","subject":"Merge \"New repo for ranger-tempest-plugin\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/fb8569a8d07810804bf51809386dce6ebab9e0ca"}]}],"author":{"name":"Clark Boylan","email":"clark.boylan@gmail.com","date":"2019-07-12 20:00:51.000000000","tz":-420},"committer":{"name":"Clark Boylan","email":"clark.boylan@gmail.com","date":"2019-07-12 20:00:51.000000000","tz":-420},"subject":"Reenable limestone cloud region in nodepool","message":"Reenable limestone cloud region in nodepool\n\nThe limestone mirror\u0027s cacti graphs don\u0027t show any of the network\ninterruptions (gaps in graphs) for about 2 weeks now. Lets try turning\nit back on.\n\nChange-Id: I00c9f644eab73ef8b4189de4ea00891dca2c8f8c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/553e143b728f493dc03b30d330f7cce9a04eaf13"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/553e143b728f493dc03b30d330f7cce9a04eaf13"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
