)]}'
{"id":"openstack%2Fproject-config~670778","triplet_id":"openstack%2Fproject-config~master~Ib9bbf91ad49f613907429e0feb4bddef1a7efa4a","project":"openstack/project-config","branch":"master","topic":"kayobe-review-priority","hashtags":[],"change_id":"Ib9bbf91ad49f613907429e0feb4bddef1a7efa4a","subject":"Add review priority for kayobe repos","status":"MERGED","created":"2019-07-15 08:08:58.000000000","updated":"2019-07-19 18:18:32.000000000","submitted":"2019-07-19 18:18:31.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"670778-1563560312274-f7d455a8","meta_rev_id":"5d8329fa5287e38984bbd6a3414f9518ed2acff0","_number":670778,"virtual_id_number":670778,"owner":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":2,"date":"2019-07-19 18:18:31.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"all":[{"value":2,"date":"2019-07-15 16:40:09.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2019-07-19 18:03:07.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"all":[{"value":0,"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2019-07-19 18:03:07.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-07-15 16:40:09.000000000","updated_by":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"reviewer":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"state":"REVIEWER"},{"updated":"2019-07-19 18:03:07.000000000","updated_by":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"reviewer":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"state":"REVIEWER"},{"updated":"2019-07-19 18:18:31.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"669c68b48a1ddc25f6f35a68f83ec09aa3582fb6","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2019-07-15 08:08:58.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"6ec79d4af5e7c6fe7d04effbbcc6ed2f7fb06a7c","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-15 08:17:20.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-zuul-jobs-linters http://logs.openstack.org/78/670778/1/check/openstack-zuul-jobs-linters/e73f5ff/ : SUCCESS in 6m 49s\n- project-config-gerrit http://logs.openstack.org/78/670778/1/check/project-config-gerrit/26df36d/ : FAILURE in 5m 00s","accounts_in_message":[],"_revision_number":1},{"id":"bffcb30b03662f592ced1813c69e0c4e27ab04b6","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2019-07-15 08:25:11.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"11d0ca0cdf7419fa56c5d1b13f4ad55ed47524c9","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-15 08:51:55.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-zuul-jobs-linters http://logs.openstack.org/78/670778/2/check/openstack-zuul-jobs-linters/3b97424/ : SUCCESS in 7m 29s\n- project-config-gerrit http://logs.openstack.org/78/670778/2/check/project-config-gerrit/ce1a6bb/ : SUCCESS in 5m 15s","accounts_in_message":[],"_revision_number":2},{"id":"f19d80159c295aa4fc4c0728295a53f4a69bf241","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2019-07-15 16:30:08.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"a16ac917dac88f026d644ac07d6a7cdaf947acd0","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2019-07-15 16:36:00.000000000","message":"Patch Set 3: Published edit on patch set 2.","accounts_in_message":[],"_revision_number":3},{"id":"76d2b6c9c2437aaeba4a0c5052a5f2aed0e24afb","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2019-07-15 16:36:28.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"c5cc902daceb494bf1a13448368aa3474b1d2a1a","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2019-07-15 16:40:09.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"a290216b5ce20491a302514decd165f56169e190","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-15 16:49:32.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-zuul-jobs-linters http://logs.openstack.org/78/670778/3/check/openstack-zuul-jobs-linters/b2aaf90/ : SUCCESS in 6m 51s\n- project-config-gerrit http://logs.openstack.org/78/670778/3/check/project-config-gerrit/5fa3afa/ : SUCCESS in 3m 41s","accounts_in_message":[],"_revision_number":3},{"id":"e2af56b50aac5aacd4b951b4e7bf0771a889d94d","author":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"date":"2019-07-19 18:03:07.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"db20409b829886c7786dcaaf878ced27ef81d55c","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-19 18:03:23.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"e891137cd2841f9b660edaaeeef52694f7f7110f","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-19 18:18:31.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-zuul-jobs-linters http://logs.openstack.org/78/670778/3/gate/openstack-zuul-jobs-linters/d721784/ : SUCCESS in 6m 42s\n- project-config-gerrit http://logs.openstack.org/78/670778/3/gate/project-config-gerrit/901b8ce/ : SUCCESS in 4m 28s","accounts_in_message":[],"_revision_number":3},{"id":"63bd9db20ee6b93852c0b7a7fdda932814979801","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-19 18:18:32.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"a722b04a38f8b5eaa1338aa7b0c3a24133358ddf","revisions":{"083fae8d435bbf6a94a815b18e056615c2b44814":{"kind":"REWORK","_number":1,"created":"2019-07-15 08:08:58.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/78/670778/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/78/670778/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/78/670778/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/78/670778/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/78/670778/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/78/670778/1"}}},"commit":{"parents":[{"commit":"4a91e22412605160114cc5c26ee124e08dba6975","subject":"Merge \"Bringing Fort Nebula CI Cloud back online\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/4a91e22412605160114cc5c26ee124e08dba6975"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2019-07-15 08:08:12.000000000","tz":60},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2019-07-15 08:08:12.000000000","tz":60},"subject":"Add review priority for kayobe repos","message":"Add review priority for kayobe repos\n\nThis brings kayobe into line with other kolla deliverables.\n\nChange-Id: Ib9bbf91ad49f613907429e0feb4bddef1a7efa4a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/083fae8d435bbf6a94a815b18e056615c2b44814"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/083fae8d435bbf6a94a815b18e056615c2b44814"}]},"branch":"refs/heads/master"},"401d6cf8e592e22ec6eb0fa5db97a85ed474ece5":{"kind":"REWORK","_number":2,"created":"2019-07-15 08:25:11.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/78/670778/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/78/670778/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/78/670778/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/78/670778/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/78/670778/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/78/670778/2"}}},"commit":{"parents":[{"commit":"4a91e22412605160114cc5c26ee124e08dba6975","subject":"Merge \"Bringing Fort Nebula CI Cloud back online\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/4a91e22412605160114cc5c26ee124e08dba6975"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2019-07-15 08:08:12.000000000","tz":60},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2019-07-15 08:24:57.000000000","tz":60},"subject":"Add review priority for kayobe repos","message":"Add review priority for kayobe repos\n\nThis brings kayobe into line with other kolla deliverables.\n\nChange-Id: Ib9bbf91ad49f613907429e0feb4bddef1a7efa4a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/401d6cf8e592e22ec6eb0fa5db97a85ed474ece5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/401d6cf8e592e22ec6eb0fa5db97a85ed474ece5"}]},"branch":"refs/heads/master"},"a722b04a38f8b5eaa1338aa7b0c3a24133358ddf":{"kind":"REWORK","_number":3,"created":"2019-07-15 16:36:00.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/78/670778/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/78/670778/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/78/670778/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/78/670778/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/78/670778/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/78/670778/3"}}},"commit":{"parents":[{"commit":"4a91e22412605160114cc5c26ee124e08dba6975","subject":"Merge \"Bringing Fort Nebula CI Cloud back online\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/4a91e22412605160114cc5c26ee124e08dba6975"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2019-07-15 08:08:12.000000000","tz":60},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2019-07-15 16:35:50.000000000","tz":0},"subject":"Add review priority for kayobe repos","message":"Add review priority for kayobe repos\n\nThis brings kayobe into line with other kolla deliverables.\n\nChange-Id: Ib9bbf91ad49f613907429e0feb4bddef1a7efa4a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/a722b04a38f8b5eaa1338aa7b0c3a24133358ddf"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/a722b04a38f8b5eaa1338aa7b0c3a24133358ddf"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
