)]}'
{"id":"openstack%2Fproject-config~670923","triplet_id":"openstack%2Fproject-config~master~I2a8ad8f8303bdd78389f62f1d8b4264698d46245","project":"openstack/project-config","branch":"master","topic":"end-airship-utils-gating","hashtags":[],"change_id":"I2a8ad8f8303bdd78389f62f1d8b4264698d46245","subject":"End Airship Utils gating","status":"MERGED","created":"2019-07-15 22:54:06.000000000","updated":"2019-07-19 17:18:03.000000000","submitted":"2019-07-19 17:18:03.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"670923-1563556683593-6d7173bf","meta_rev_id":"188059b488c95a03da8f20e0c570ceeb6176d803","_number":670923,"virtual_id_number":670923,"owner":{"_account_id":22477,"name":"Matt McEuen","email":"matt.mceuen@att.com","username":"mattmceuen"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"value":0,"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":2,"date":"2019-07-19 17:18:03.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"date":"2019-07-18 18:53:57.000000000","_account_id":22477,"name":"Matt McEuen","email":"matt.mceuen@att.com","username":"mattmceuen"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"all":[{"value":2,"date":"2019-07-19 17:04:01.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"value":2,"date":"2019-07-18 18:56:15.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":22477,"name":"Matt McEuen","email":"matt.mceuen@att.com","username":"mattmceuen"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"all":[{"value":1,"date":"2019-07-19 17:04:01.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"value":0,"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":22477,"name":"Matt McEuen","email":"matt.mceuen@att.com","username":"mattmceuen"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22477,"name":"Matt McEuen","email":"matt.mceuen@att.com","username":"mattmceuen"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-07-18 18:56:15.000000000","updated_by":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"reviewer":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"state":"REVIEWER"},{"updated":"2019-07-19 17:04:01.000000000","updated_by":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"reviewer":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"state":"REVIEWER"},{"updated":"2019-07-19 17:18:03.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"d798c4217416b5ce28c52279f86a8aaca5b5cd0a","author":{"_account_id":22477,"name":"Matt McEuen","email":"matt.mceuen@att.com","username":"mattmceuen"},"date":"2019-07-15 22:54:06.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"aeb9e20bdf4c916472468550665185fe52c340e9","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-15 23:02:17.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-zuul-jobs-linters http://logs.openstack.org/23/670923/1/check/openstack-zuul-jobs-linters/6a4f8ef/ : SUCCESS in 6m 07s","accounts_in_message":[],"_revision_number":1},{"id":"86782fbe1ec5d5eb7f4d35d1a0f5ab13d572cb5d","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2019-07-16 04:23:50.000000000","message":"Patch Set 1:\n\nAre you following https://docs.openstack.org/infra/manual/drivers.html#retiring-a-project ? Did you announce this anywhere?\n\nI\u0027d like to see *all* patches up before we merge this, could you do so (even if some fail right now), please?","accounts_in_message":[],"_revision_number":1},{"id":"3c0260bb37dffe94be9828293eb2334601d5a5fd","author":{"_account_id":22477,"name":"Matt McEuen","email":"matt.mceuen@att.com","username":"mattmceuen"},"date":"2019-07-16 13:41:22.000000000","message":"Patch Set 1:\n\nHi Andreas:  yes, I\u0027m following those instructions, and this was announced/discussed at the PTG and a couple times in the IRC meeting.  I can definitely put all the patches up; was just holding off because I interpreted the manual to say they should be sequenced.  Thanks!","accounts_in_message":[],"_revision_number":1},{"id":"a5976bbc2d3ef80d9c2729752c5cb254b94b5b2a","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2019-07-16 13:48:20.000000000","message":"Patch Set 1:\n\nThose are separate steps (reviews) but you can put them out at the same time","accounts_in_message":[],"_revision_number":1},{"id":"fc4b13cd5cf64b80c801e697130279e24d1aaf36","author":{"_account_id":22477,"name":"Matt McEuen","email":"matt.mceuen@att.com","username":"mattmceuen"},"date":"2019-07-18 18:53:57.000000000","message":"Patch Set 1:\n\nThe change to zero-out the project is here:\nhttps://review.opendev.org/#/c/671042/\n\nThe change to remove the project from infra is here:\nhttps://review.opendev.org/#/c/671547/\n\nPlease let me know if I mangled anything.  Thanks!","accounts_in_message":[],"_revision_number":1},{"id":"04f602707008bf2e51bb1b65568d260cd85d6f56","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2019-07-18 18:56:15.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"3469312c0af4e32099a1cb36eb8af96ba0b07ea7","author":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"date":"2019-07-19 17:04:01.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"a65d0192c312dcf64fcb308c8338ee4de58c354f","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-19 17:04:19.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"2409952f86068095523e58de41a56eaffa372a55","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-19 17:18:03.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-zuul-jobs-linters http://logs.openstack.org/23/670923/1/gate/openstack-zuul-jobs-linters/733f9b0/ : SUCCESS in 7m 23s","accounts_in_message":[],"_revision_number":1},{"id":"98eabd28a95bf5b5f107f86c39753973cf72410d","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-19 17:18:03.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"ce1a7d2450815710bd5dac140ccd3eb45a0686b3","revisions":{"ce1a7d2450815710bd5dac140ccd3eb45a0686b3":{"kind":"REWORK","_number":1,"created":"2019-07-15 22:54:06.000000000","uploader":{"_account_id":22477,"name":"Matt McEuen","email":"matt.mceuen@att.com","username":"mattmceuen"},"ref":"refs/changes/23/670923/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/23/670923/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/23/670923/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/23/670923/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/23/670923/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/23/670923/1"}}},"commit":{"parents":[{"commit":"6924bb9a48d1cd8cb2f1ddaffeeaa2a44cb12457","subject":"Normalize projects.yaml","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/6924bb9a48d1cd8cb2f1ddaffeeaa2a44cb12457"}]}],"author":{"name":"Matt McEuen","email":"matt.mceuen@att.com","date":"2019-07-15 22:50:56.000000000","tz":-300},"committer":{"name":"Matt McEuen","email":"matt.mceuen@att.com","date":"2019-07-15 22:54:05.000000000","tz":-300},"subject":"End Airship Utils gating","message":"End Airship Utils gating\n\nThis ends airship/utils gating in advance of retiring this project.\n\nChange-Id: I2a8ad8f8303bdd78389f62f1d8b4264698d46245\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/ce1a7d2450815710bd5dac140ccd3eb45a0686b3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/ce1a7d2450815710bd5dac140ccd3eb45a0686b3"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
