)]}'
{"id":"openstack%2Fproject-config~671268","triplet_id":"openstack%2Fproject-config~master~I28c86d332fd0c63f854ce8cf6eded927e3584d3a","project":"openstack/project-config","branch":"master","topic":"retire-puppet-crane","hashtags":[],"change_id":"I28c86d332fd0c63f854ce8cf6eded927e3584d3a","subject":"Remove puppet-crane","status":"MERGED","created":"2019-07-17 13:23:43.000000000","updated":"2019-07-26 14:32:00.000000000","submitted":"2019-07-26 14:32:00.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"671268-1564151520667-a68571c6","meta_rev_id":"4dfaf6e8ffd5c00ba06c8d1394f1dd83f17b0387","_number":671268,"virtual_id_number":671268,"owner":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},{"value":0,"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":2,"date":"2019-07-26 14:32:00.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},"all":[{"value":2,"date":"2019-07-26 14:06:30.000000000","_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},{"value":2,"date":"2019-07-17 13:57:29.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},"all":[{"value":1,"date":"2019-07-26 14:06:30.000000000","_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},{"value":0,"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-07-17 13:57:29.000000000","updated_by":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"reviewer":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"state":"REVIEWER"},{"updated":"2019-07-26 14:06:30.000000000","updated_by":{"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},"reviewer":{"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},"state":"REVIEWER"},{"updated":"2019-07-26 14:32:00.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"19860034ce711f4c110362a2b69d890bcdc844a1","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2019-07-17 13:23:43.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"3253464edff89e29c04bbc9f55fd95262d9c5e86","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2019-07-17 13:34:34.000000000","message":"Uploaded patch set 2: Patch Set 1 was rebased.","accounts_in_message":[],"_revision_number":2},{"id":"377571c320883814c3a413c0d1a5bc22bcf5b5e3","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-17 13:34:45.000000000","message":"Patch Set 1: Verified-1\n\nThis change depends on a change that failed to merge.","accounts_in_message":[],"_revision_number":1},{"id":"7e54e523c4dcaee48513db1af77b68cbf88146ef","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2019-07-17 13:47:32.000000000","message":"Patch Set 2:\n\nLGTM - do you have a governance change up as well?","accounts_in_message":[],"_revision_number":2},{"id":"de859e11b154c9b442fd2ca426de576fe2847dc1","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2019-07-17 13:49:22.000000000","message":"Patch Set 2:\n\nyea they renamed the topic so it\u0027s hard to find now https://review.opendev.org/#/c/671271/","accounts_in_message":[],"_revision_number":2},{"id":"095accea37450fdac88eec972b027216d72e9884","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-17 13:50:10.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-zuul-jobs-linters http://logs.openstack.org/68/671268/2/check/openstack-zuul-jobs-linters/eb114fa/ : SUCCESS in 7m 46s\n- project-config-gerrit http://logs.openstack.org/68/671268/2/check/project-config-gerrit/01c09b0/ : SUCCESS in 3m 29s\n- project-config-irc-access http://logs.openstack.org/68/671268/2/check/project-config-irc-access/7cfcd6c/ : SUCCESS in 6m 28s","accounts_in_message":[],"_revision_number":2},{"id":"e3a0ffdc36edff01c07698a2a4f5d350627f8f05","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2019-07-17 13:51:59.000000000","message":"Patch Set 2: Code-Review+2\n\nyou could add to commit message:\n\nNeeded-By: https://review.opendev.org/#/c/671271/ \n\nThat makes it easy to find ;)","accounts_in_message":[],"_revision_number":2},{"id":"93f608cdf9faebe6df7fba4a50fbe9ed2c87c370","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2019-07-17 13:56:40.000000000","message":"Patch Set 2:\n\nsure except i posted the other one before this one :D but i\u0027ll update it","accounts_in_message":[],"_revision_number":2},{"id":"b62bf88e8caa6badd6c50c8ee9330949bd77c377","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2019-07-17 13:56:54.000000000","message":"Patch Set 2:\n\ner this one before the other one.","accounts_in_message":[],"_revision_number":2},{"id":"2473dc72b7bfa6c5d6cb10ac7619e3890f0fbe3f","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2019-07-17 13:57:04.000000000","message":"Patch Set 3: Commit message was updated.","accounts_in_message":[],"_revision_number":3},{"id":"c61d1e8d7f69662f419184196896bfa7e3712e9a","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2019-07-17 13:57:29.000000000","message":"Patch Set 3: Code-Review+2\n\nthanks","accounts_in_message":[],"_revision_number":3},{"id":"f493dcb25ac92c8649d878476e3b01e02d32bc6d","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-17 14:11:46.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-zuul-jobs-linters http://logs.openstack.org/68/671268/3/check/openstack-zuul-jobs-linters/e6f87ea/ : SUCCESS in 7m 03s\n- project-config-gerrit http://logs.openstack.org/68/671268/3/check/project-config-gerrit/f7eb7a2/ : SUCCESS in 5m 10s\n- project-config-irc-access http://logs.openstack.org/68/671268/3/check/project-config-irc-access/17069f8/ : SUCCESS in 6m 40s","accounts_in_message":[],"_revision_number":3},{"id":"1122600917210be7b922c1323640ae8727670fcb","author":{"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},"date":"2019-07-26 14:06:30.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"d33e54a718135e20ff6bfc9e208f19d583c0f83e","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-26 14:06:49.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"b5e3ca2316223924269b486ebd27192e2717efec","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-26 14:32:00.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":3},{"id":"91768d07e356096f7054d43a3fe584d026a762e5","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-26 14:32:00.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-zuul-jobs-linters http://logs.openstack.org/68/671268/3/gate/openstack-zuul-jobs-linters/b74a1b8/ : SUCCESS in 8m 33s\n- project-config-gerrit http://logs.openstack.org/68/671268/3/gate/project-config-gerrit/14daa9c/ : SUCCESS in 3m 37s\n- project-config-irc-access http://logs.openstack.org/68/671268/3/gate/project-config-irc-access/d618299/ : SUCCESS in 6m 06s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"e2973c30e8224ca38db8413386ddb44163ae8aad","revisions":{"76afe114e6d39b24c3982b667e4cf1c632cbc77a":{"kind":"REWORK","_number":1,"created":"2019-07-17 13:23:43.000000000","uploader":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"ref":"refs/changes/68/671268/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/68/671268/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/68/671268/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/68/671268/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/68/671268/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/68/671268/1"}}},"commit":{"parents":[{"commit":"b200bdce34e3dd192e87415eb06c5f6f3aa8057d","subject":"Add puppet-crane noop jobs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/b200bdce34e3dd192e87415eb06c5f6f3aa8057d"}]}],"author":{"name":"Alex Schultz","email":"aschultz@redhat.com","date":"2019-07-17 13:23:20.000000000","tz":-360},"committer":{"name":"Alex Schultz","email":"aschultz@redhat.com","date":"2019-07-17 13:23:20.000000000","tz":-360},"subject":"Remove puppet-crane","message":"Remove puppet-crane\n\nDepends-On: https://review.opendev.org/#/c/671267/\nChange-Id: I28c86d332fd0c63f854ce8cf6eded927e3584d3a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/76afe114e6d39b24c3982b667e4cf1c632cbc77a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/76afe114e6d39b24c3982b667e4cf1c632cbc77a"}]},"branch":"refs/heads/master"},"6c3a36f1f070f3debca07048104635f962193858":{"kind":"REWORK","_number":2,"created":"2019-07-17 13:34:34.000000000","uploader":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"ref":"refs/changes/68/671268/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/68/671268/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/68/671268/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/68/671268/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/68/671268/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/68/671268/2"}}},"commit":{"parents":[{"commit":"4a251c568191e8610824f8f274cf243de72f0710","subject":"Add puppet-crane noop jobs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/4a251c568191e8610824f8f274cf243de72f0710"}]}],"author":{"name":"Alex Schultz","email":"aschultz@redhat.com","date":"2019-07-17 13:23:20.000000000","tz":-360},"committer":{"name":"Alex Schultz","email":"aschultz@redhat.com","date":"2019-07-17 13:34:25.000000000","tz":-360},"subject":"Remove puppet-crane","message":"Remove puppet-crane\n\nDepends-On: https://review.opendev.org/#/c/671267/\nChange-Id: I28c86d332fd0c63f854ce8cf6eded927e3584d3a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/6c3a36f1f070f3debca07048104635f962193858"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/6c3a36f1f070f3debca07048104635f962193858"}]},"branch":"refs/heads/master"},"e2973c30e8224ca38db8413386ddb44163ae8aad":{"kind":"NO_CODE_CHANGE","_number":3,"created":"2019-07-17 13:57:04.000000000","uploader":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"ref":"refs/changes/68/671268/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/68/671268/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/68/671268/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/68/671268/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/68/671268/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/68/671268/3"}}},"commit":{"parents":[{"commit":"4a251c568191e8610824f8f274cf243de72f0710","subject":"Add puppet-crane noop jobs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/4a251c568191e8610824f8f274cf243de72f0710"}]}],"author":{"name":"Alex Schultz","email":"aschultz@redhat.com","date":"2019-07-17 13:23:20.000000000","tz":-360},"committer":{"name":"Alex Schultz","email":"aschultz@redhat.com","date":"2019-07-17 13:57:02.000000000","tz":0},"subject":"Remove puppet-crane","message":"Remove puppet-crane\n\nDepends-On: https://review.opendev.org/#/c/671267/\nNeeded-By: https://review.opendev.org/#/c/671271/\nChange-Id: I28c86d332fd0c63f854ce8cf6eded927e3584d3a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/e2973c30e8224ca38db8413386ddb44163ae8aad"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/e2973c30e8224ca38db8413386ddb44163ae8aad"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
