)]}'
{"id":"openstack%2Fproject-config~671483","triplet_id":"openstack%2Fproject-config~master~I740c66cd959db1efa8ca8c9f29d4da4b66fbd993","project":"openstack/project-config","branch":"master","topic":"retire-doc8","hashtags":[],"change_id":"I740c66cd959db1efa8ca8c9f29d4da4b66fbd993","subject":"Ends project gating for doc8","status":"MERGED","created":"2019-07-18 13:25:17.000000000","updated":"2019-07-24 11:42:51.000000000","submitted":"2019-07-24 11:42:51.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"671483-1563968571540-53055c7d","meta_rev_id":"4dabb1d7c330175155055eb4f9c3e7e7ae169067","_number":671483,"virtual_id_number":671483,"owner":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},{"value":0,"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":2,"date":"2019-07-24 11:42:51.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},"all":[{"value":2,"date":"2019-07-24 11:31:18.000000000","_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},{"value":0,"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},"all":[{"value":1,"date":"2019-07-24 11:31:18.000000000","_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},{"value":0,"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-07-18 14:29:06.000000000","updated_by":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"reviewer":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"state":"REVIEWER"},{"updated":"2019-07-24 11:31:18.000000000","updated_by":{"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},"reviewer":{"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},"state":"REVIEWER"},{"updated":"2019-07-24 11:42:51.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"aa1ce40abd13affffc3033c8311d4c7288a663c6","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2019-07-18 13:25:17.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"b683d0032b010088be0bb601fc7453efbbe3cd8a","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2019-07-18 13:26:33.000000000","message":"Patch Set 1: Workflow-1\n\nI\u0027d really like to merge two more patch against \u0027x/doc8\u0027 before merging this, so I\u0027m going to -W for a bit in the hope harlowja responds to my core request. If not, I\u0027ll drop the -W and we can fix things up on GitHub.","accounts_in_message":[],"_revision_number":1},{"id":"9141e67a4c8166b198ee5b59ab6b8da87cd5ec78","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2019-07-18 13:27:36.000000000","message":"Topic changed from pycqa-migrate to retire-doc8","accounts_in_message":[],"_revision_number":1},{"id":"065a572ec39449a9c456cf8da939219e89eeb602","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-18 13:41:48.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-zuul-jobs-linters http://logs.openstack.org/83/671483/1/check/openstack-zuul-jobs-linters/e40b59b/ : SUCCESS in 6m 30s","accounts_in_message":[],"_revision_number":1},{"id":"bb6fb4f9a6dab942e03d3b55b92304c31f0cefe8","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2019-07-18 14:29:06.000000000","message":"Patch Set 1: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"96b22164adde37c9ab3bc59bb9893d9318ee04c8","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2019-07-19 15:01:15.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"2ff4d8d7b37b5d5f58a5b50b38503ac807b3cfca","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2019-07-24 09:16:53.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"035231d1e177074baad6b34942ba092cf9899cc2","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-24 09:28:49.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-zuul-jobs-linters http://logs.openstack.org/83/671483/2/check/openstack-zuul-jobs-linters/a023597/ : SUCCESS in 6m 23s","accounts_in_message":[],"_revision_number":2},{"id":"9e3213ec7e5c133aa844d9239d8b9cddb20197a0","author":{"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},"date":"2019-07-24 11:31:18.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"83648cc90a8f0ccaff7bb8257bfba1911489860d","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-24 11:31:36.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"a84e2bfcbd7dc9abbbc68d89204ed15cf4e25fc5","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-24 11:42:51.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"4dabb1d7c330175155055eb4f9c3e7e7ae169067","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-24 11:42:51.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-zuul-jobs-linters http://logs.openstack.org/83/671483/2/gate/openstack-zuul-jobs-linters/3c682da/ : SUCCESS in 9m 58s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"368231b5ada9e6e9b7bbc25feaad553434a7a3e8","revisions":{"ce597705739274bc9466fa05f9667d426f14f041":{"kind":"REWORK","_number":1,"created":"2019-07-18 13:25:17.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/83/671483/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/83/671483/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/83/671483/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/83/671483/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/83/671483/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/83/671483/1"}}},"commit":{"parents":[{"commit":"465a6a63b01c65284e64cb4f70683284d385bb45","subject":"Normalize projects.yaml","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/465a6a63b01c65284e64cb4f70683284d385bb45"}]}],"author":{"name":"Stephen Finucane","email":"sfinucan@redhat.com","date":"2019-07-18 13:08:19.000000000","tz":60},"committer":{"name":"Stephen Finucane","email":"sfinucan@redhat.com","date":"2019-07-18 13:08:19.000000000","tz":60},"subject":"Ends project gating for Bandit","message":"Ends project gating for Bandit\n\nRemove doc8 from Gate, as the project is being migrated to PyCQA [0].\n\n[0] http://lists.openstack.org/pipermail/openstack-discuss/2019-July/007878.html\n\nChange-Id: I740c66cd959db1efa8ca8c9f29d4da4b66fbd993\nSigned-off-by: Stephen Finucane \u003csfinucan@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/ce597705739274bc9466fa05f9667d426f14f041"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/ce597705739274bc9466fa05f9667d426f14f041"}]},"branch":"refs/heads/master"},"368231b5ada9e6e9b7bbc25feaad553434a7a3e8":{"kind":"TRIVIAL_REBASE_WITH_MESSAGE_UPDATE","_number":2,"created":"2019-07-24 09:16:53.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/83/671483/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/83/671483/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/83/671483/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/83/671483/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/83/671483/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/83/671483/2"}}},"commit":{"parents":[{"commit":"27d5e9808a33f03767b74486e2bf3f586a8ce58a","subject":"Add additional gerrit plugin repos","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/27d5e9808a33f03767b74486e2bf3f586a8ce58a"}]}],"author":{"name":"Stephen Finucane","email":"sfinucan@redhat.com","date":"2019-07-18 13:08:19.000000000","tz":60},"committer":{"name":"Stephen Finucane","email":"sfinucan@redhat.com","date":"2019-07-24 09:16:14.000000000","tz":60},"subject":"Ends project gating for doc8","message":"Ends project gating for doc8\n\nRemove doc8 from the gate as the project has been migrated to PyCQA [1].\n\n[1] http://lists.openstack.org/pipermail/openstack-discuss/2019-July/007878.html\n\nChange-Id: I740c66cd959db1efa8ca8c9f29d4da4b66fbd993\nSigned-off-by: Stephen Finucane \u003csfinucan@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/368231b5ada9e6e9b7bbc25feaad553434a7a3e8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/368231b5ada9e6e9b7bbc25feaad553434a7a3e8"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
