)]}'
{"id":"openstack%2Fproject-config~671550","triplet_id":"openstack%2Fproject-config~master~Ibd6c51710bce82baa79d6733064f69f3f0e2c8e8","project":"openstack/project-config","branch":"master","topic":"remove-airship-berth","hashtags":[],"change_id":"Ibd6c51710bce82baa79d6733064f69f3f0e2c8e8","subject":"Remove airship/berth from infra","status":"MERGED","created":"2019-07-18 18:58:48.000000000","updated":"2019-07-19 17:56:07.000000000","submitted":"2019-07-19 17:56:07.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":4,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"671550-1563558967975-77280bbb","meta_rev_id":"7bbd9d89d25cd8979f359bf6108760f1d06fffb7","_number":671550,"virtual_id_number":671550,"owner":{"_account_id":22477,"name":"Matt McEuen","email":"matt.mceuen@att.com","username":"mattmceuen"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"value":0,"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":2,"date":"2019-07-19 17:56:07.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":22477,"name":"Matt McEuen","email":"matt.mceuen@att.com","username":"mattmceuen"},{"value":0,"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"all":[{"value":0,"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"value":2,"date":"2019-07-19 17:32:22.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":22477,"name":"Matt McEuen","email":"matt.mceuen@att.com","username":"mattmceuen"},{"value":2,"date":"2019-07-19 17:36:12.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"all":[{"value":0,"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"value":0,"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":22477,"name":"Matt McEuen","email":"matt.mceuen@att.com","username":"mattmceuen"},{"value":1,"date":"2019-07-19 17:36:12.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22477,"name":"Matt McEuen","email":"matt.mceuen@att.com","username":"mattmceuen"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-07-19 17:04:34.000000000","updated_by":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"reviewer":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"state":"REVIEWER"},{"updated":"2019-07-19 17:32:22.000000000","updated_by":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"reviewer":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"state":"REVIEWER"},{"updated":"2019-07-19 17:36:12.000000000","updated_by":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"reviewer":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"state":"REVIEWER"},{"updated":"2019-07-19 17:56:07.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"f69d81e31059c4c16343eaed152d86113d2b8103","author":{"_account_id":22477,"name":"Matt McEuen","email":"matt.mceuen@att.com","username":"mattmceuen"},"date":"2019-07-18 18:58:48.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"bbad07f2a2e337cfac5d9480b69b6aa0b6021249","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-18 19:09:12.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-zuul-jobs-linters http://logs.openstack.org/50/671550/1/check/openstack-zuul-jobs-linters/7b45a2e/ : SUCCESS in 8m 45s\n- project-config-gerrit http://logs.openstack.org/50/671550/1/check/project-config-gerrit/ed4c64d/ : SUCCESS in 3m 24s","accounts_in_message":[],"_revision_number":1},{"id":"4dc54d36f2ad43a73b7b98bdc4c3b10da940a5a7","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2019-07-18 19:18:27.000000000","message":"Patch Set 1: Code-Review-1\n\n(2 comments)","accounts_in_message":[],"_revision_number":1},{"id":"6b0c967435fa93771521610e87ea3c39a382a298","author":{"_account_id":22477,"name":"Matt McEuen","email":"matt.mceuen@att.com","username":"mattmceuen"},"date":"2019-07-18 22:00:37.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"145d4ed288750c94a663f1cf4897882bdfb36fe9","author":{"_account_id":22477,"name":"Matt McEuen","email":"matt.mceuen@att.com","username":"mattmceuen"},"date":"2019-07-18 22:02:38.000000000","message":"Patch Set 1:\n\n(2 comments)","accounts_in_message":[],"_revision_number":1},{"id":"4855171c47afa1437d2d3c3ba7c15b6bb1a51bd2","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-18 22:08:39.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-zuul-jobs-linters http://logs.openstack.org/50/671550/2/check/openstack-zuul-jobs-linters/cb4bd83/ : SUCCESS in 6m 33s\n- project-config-gerrit http://logs.openstack.org/50/671550/2/check/project-config-gerrit/ff988ad/ : SUCCESS in 3m 55s","accounts_in_message":[],"_revision_number":2},{"id":"1ae74272d51c177ccf86304c9d82343bdf4e7d1d","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2019-07-19 06:38:43.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"cf406d053d9c60f86fc2ef6c23635e608757d6e1","author":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"date":"2019-07-19 17:04:34.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"8d9fb808a0970b34c4e5c3009f52a59abe82cea7","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-19 17:18:05.000000000","message":"Patch Set 2: Verified-2\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.","accounts_in_message":[],"_revision_number":2},{"id":"8d271425dda95e648be3c28362d96d6300150c70","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2019-07-19 17:31:51.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"584a09d5410fde8279e5551ae3c2b1daffb94a32","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2019-07-19 17:32:22.000000000","message":"Patch Set 3: Code-Review+2\n\nrebased to resolve merge conflict","accounts_in_message":[],"_revision_number":3},{"id":"f155cb0b78b2d4dc59601fb31e8066f88e479a96","author":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"date":"2019-07-19 17:36:12.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"2b702d27ef351867df7eaf0e4a628f85a20994dd","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-19 17:42:44.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-zuul-jobs-linters http://logs.openstack.org/50/671550/3/check/openstack-zuul-jobs-linters/48402bb/ : SUCCESS in 7m 28s\n- project-config-gerrit http://logs.openstack.org/50/671550/3/check/project-config-gerrit/e47acd0/ : SUCCESS in 4m 28s","accounts_in_message":[],"_revision_number":3},{"id":"e6872aeeb213089ebaeb6fb9d9da4f0d6a9ebc5f","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-19 17:43:03.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"eb1853ced489f411649d5ecce0a603dfde8af75b","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-19 17:56:07.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":3},{"id":"1718f63d1400593f34b3d1661d8ef30a07c46ade","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-19 17:56:07.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-zuul-jobs-linters http://logs.openstack.org/50/671550/3/gate/openstack-zuul-jobs-linters/7a313b5/ : SUCCESS in 8m 07s\n- project-config-gerrit http://logs.openstack.org/50/671550/3/gate/project-config-gerrit/5ff76e7/ : SUCCESS in 4m 22s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"f71a02934374c64884b9b15743359603d6aed4ea","revisions":{"402d3b0e44a9b30f54171cc1c15cff1a87d7747a":{"kind":"REWORK","_number":1,"created":"2019-07-18 18:58:48.000000000","uploader":{"_account_id":22477,"name":"Matt McEuen","email":"matt.mceuen@att.com","username":"mattmceuen"},"ref":"refs/changes/50/671550/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/50/671550/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/50/671550/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/50/671550/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/50/671550/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/50/671550/1"}}},"commit":{"parents":[{"commit":"465a6a63b01c65284e64cb4f70683284d385bb45","subject":"Normalize projects.yaml","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/465a6a63b01c65284e64cb4f70683284d385bb45"}]}],"author":{"name":"Matt McEuen","email":"matt.mceuen@att.com","date":"2019-07-18 18:57:36.000000000","tz":-300},"committer":{"name":"Matt McEuen","email":"matt.mceuen@att.com","date":"2019-07-18 18:57:36.000000000","tz":-300},"subject":"Remove airship/berth from infra","message":"Remove airship/berth from infra\n\nThis removes the airship/berth project from OpenStack Infra,\nas agreed upon at the Denver PTG and in subsequent Airship\nIRC team meetings.\n\nThe PS to zero-out the project has already merged:\nhttps://review.opendev.org/#/c/670924/\n\nChange-Id: Ibd6c51710bce82baa79d6733064f69f3f0e2c8e8\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/402d3b0e44a9b30f54171cc1c15cff1a87d7747a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/402d3b0e44a9b30f54171cc1c15cff1a87d7747a"}]},"branch":"refs/heads/master"},"a237948a8222fcf349dd0bb317d475d7d5cc1b54":{"kind":"REWORK","_number":2,"created":"2019-07-18 22:00:37.000000000","uploader":{"_account_id":22477,"name":"Matt McEuen","email":"matt.mceuen@att.com","username":"mattmceuen"},"ref":"refs/changes/50/671550/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/50/671550/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/50/671550/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/50/671550/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/50/671550/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/50/671550/2"}}},"commit":{"parents":[{"commit":"465a6a63b01c65284e64cb4f70683284d385bb45","subject":"Normalize projects.yaml","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/465a6a63b01c65284e64cb4f70683284d385bb45"}]}],"author":{"name":"Matt McEuen","email":"matt.mceuen@att.com","date":"2019-07-18 18:57:36.000000000","tz":-300},"committer":{"name":"Matt McEuen","email":"matt.mceuen@att.com","date":"2019-07-18 22:00:28.000000000","tz":-300},"subject":"Remove airship/berth from infra","message":"Remove airship/berth from infra\n\nThis removes the airship/berth project from OpenStack Infra,\nas agreed upon at the Denver PTG and in subsequent Airship\nIRC team meetings.\n\nThe PS to zero-out the project has already merged:\nhttps://review.opendev.org/#/c/670924/\n\nChange-Id: Ibd6c51710bce82baa79d6733064f69f3f0e2c8e8\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/a237948a8222fcf349dd0bb317d475d7d5cc1b54"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/a237948a8222fcf349dd0bb317d475d7d5cc1b54"}]},"branch":"refs/heads/master"},"f71a02934374c64884b9b15743359603d6aed4ea":{"kind":"REWORK","_number":3,"created":"2019-07-19 17:31:51.000000000","uploader":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"ref":"refs/changes/50/671550/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/50/671550/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/50/671550/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/50/671550/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/50/671550/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/50/671550/3"}}},"commit":{"parents":[{"commit":"c34e7bede6fc0247cf05adc99da678bd9123b32e","subject":"Merge \"Add puppet-crane noop jobs\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/c34e7bede6fc0247cf05adc99da678bd9123b32e"}]}],"author":{"name":"Matt McEuen","email":"matt.mceuen@att.com","date":"2019-07-18 18:57:36.000000000","tz":-300},"committer":{"name":"Andreas Jaeger","email":"aj@suse.com","date":"2019-07-19 17:31:12.000000000","tz":120},"subject":"Remove airship/berth from infra","message":"Remove airship/berth from infra\n\nThis removes the airship/berth project from OpenStack Infra,\nas agreed upon at the Denver PTG and in subsequent Airship\nIRC team meetings.\n\nThe PS to zero-out the project has already merged:\nhttps://review.opendev.org/#/c/670924/\n\nChange-Id: Ibd6c51710bce82baa79d6733064f69f3f0e2c8e8\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/f71a02934374c64884b9b15743359603d6aed4ea"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/f71a02934374c64884b9b15743359603d6aed4ea"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
