)]}'
{"id":"openstack%2Fproject-config~671568","triplet_id":"openstack%2Fproject-config~master~I01320361e6dc05b43f46451ef128116eb1aa3aa6","project":"openstack/project-config","branch":"master","topic":"update-neutron-dashboard","hashtags":[],"change_id":"I01320361e6dc05b43f46451ef128116eb1aa3aa6","subject":"Update Neutron grafana dashboard","status":"MERGED","created":"2019-07-18 20:40:54.000000000","updated":"2019-07-19 18:18:21.000000000","submitted":"2019-07-19 18:18:21.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"671568-1563560301385-4866acf4","meta_rev_id":"639ee16a687d2bbf8d770d2307398767e101bb21","_number":671568,"virtual_id_number":671568,"owner":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":0,"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},{"value":2,"date":"2019-07-19 18:18:20.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"all":[{"value":2,"date":"2019-07-19 06:39:22.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":2,"date":"2019-07-19 18:00:42.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"all":[{"value":0,"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":1,"date":"2019-07-19 18:00:42.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-07-19 06:39:22.000000000","updated_by":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"reviewer":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"state":"REVIEWER"},{"updated":"2019-07-19 18:00:42.000000000","updated_by":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"reviewer":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"state":"REVIEWER"},{"updated":"2019-07-19 18:18:20.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"9398d2e294872a79c8052ab9e7b550a82b163b30","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2019-07-18 20:40:54.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"6836c4981a82941808e9faa95e04aed8c9d36672","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-18 20:48:37.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-zuul-jobs-linters http://logs.openstack.org/68/671568/1/check/openstack-zuul-jobs-linters/9b298bc/ : SUCCESS in 6m 13s\n- project-config-grafyaml http://logs.openstack.org/68/671568/1/check/project-config-grafyaml/043f8b7/ : SUCCESS in 3m 07s","accounts_in_message":[],"_revision_number":1},{"id":"6a3a506f471fa6c568091b884d7b113267570a20","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2019-07-19 06:39:22.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"dfdca8659774f65667de00af5c14af9da0796070","author":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"date":"2019-07-19 18:00:42.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"3925eceecf65c6be8c910afe4efe8a0edf42294d","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-19 18:00:56.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"cbb550eeec4427eae5a975e3ad008bb26dde8e75","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-19 18:18:20.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-zuul-jobs-linters http://logs.openstack.org/68/671568/1/gate/openstack-zuul-jobs-linters/480be33/ : SUCCESS in 7m 41s\n- project-config-grafyaml http://logs.openstack.org/68/671568/1/gate/project-config-grafyaml/114b296/ : SUCCESS in 4m 02s","accounts_in_message":[],"_revision_number":1},{"id":"5a806ae50e85662f94b0ad9265632fc37efe83d9","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-19 18:18:21.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"cb3e19c60c8391fdda2ae9f8def022e921dccbae","revisions":{"cb3e19c60c8391fdda2ae9f8def022e921dccbae":{"kind":"REWORK","_number":1,"created":"2019-07-18 20:40:54.000000000","uploader":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"ref":"refs/changes/68/671568/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/68/671568/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/68/671568/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/68/671568/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/68/671568/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/68/671568/1"}}},"commit":{"parents":[{"commit":"465a6a63b01c65284e64cb4f70683284d385bb45","subject":"Normalize projects.yaml","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/465a6a63b01c65284e64cb4f70683284d385bb45"}]}],"author":{"name":"Slawek Kaplonski","email":"skaplons@redhat.com","date":"2019-07-18 20:37:39.000000000","tz":120},"committer":{"name":"Slawek Kaplonski","email":"skaplons@redhat.com","date":"2019-07-18 20:37:39.000000000","tz":120},"subject":"Update Neutron grafana dashboard","message":"Update Neutron grafana dashboard\n\nRecently there was couple of changes in Neutron CI jobs config:\n\n- [1] added new job \"tempest-integrated-networking\",\n- [2] removed tempest-full and tempest-full-py3 jobs - those jobs are\n  replaced with \"tempest-integrated-networking\" job,\n- [3] added new non-voting job:\n  tripleo-ci-centos-7-scenario007-standalone\n\nThis patch reflects above changes in Neutron Grafana dashboard.\n\n[1] https://review.opendev.org/#/c/669815/\n[2] https://review.opendev.org/#/c/670717/\n[3] https://review.opendev.org/#/c/670029/\n\nChange-Id: I01320361e6dc05b43f46451ef128116eb1aa3aa6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/cb3e19c60c8391fdda2ae9f8def022e921dccbae"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/cb3e19c60c8391fdda2ae9f8def022e921dccbae"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
