)]}'
{"id":"openstack%2Fproject-config~672290","triplet_id":"openstack%2Fproject-config~master~I571afd0557f52ec9bf470c9ede46b9e3a0bb308f","project":"openstack/project-config","branch":"master","topic":"update-neutron-dashboard","hashtags":[],"change_id":"I571afd0557f52ec9bf470c9ede46b9e3a0bb308f","subject":"Rename \"tripleo-ci-centos-7-scenario007-standalone\" in Neutron","status":"MERGED","created":"2019-07-23 13:43:09.000000000","updated":"2019-07-31 05:45:55.000000000","submitted":"2019-07-31 05:45:55.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":3,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"672290-1564551955762-cc82055b","meta_rev_id":"21fbd913f1c66a64b2ae341ddb41935f2964267e","_number":672290,"virtual_id_number":672290,"owner":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":0,"date":"2019-07-23 14:07:28.000000000","_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":2,"date":"2019-07-31 05:45:55.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"all":[{"value":2,"date":"2019-07-23 14:08:14.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2019-07-31 05:38:16.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"all":[{"value":0,"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2019-07-31 05:38:16.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-07-23 14:08:14.000000000","updated_by":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"reviewer":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"state":"REVIEWER"},{"updated":"2019-07-31 05:38:16.000000000","updated_by":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"reviewer":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"state":"REVIEWER"},{"updated":"2019-07-31 05:45:55.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"762a70825a2439ff75475ca9ca6f9a997791886f","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2019-07-23 13:43:09.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"eae4403b5690aafe327b3fa3ff684c686e5e4950","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2019-07-23 13:51:10.000000000","message":"Patch Set 1:\n\nYou should follow the naming conventions at https://docs.openstack.org/infra/manual/drivers.html#consistent-naming-for-jobs-with-zuul-v3\n\nA neutron only job should be called neutron-X","accounts_in_message":[],"_revision_number":1},{"id":"b1a1b6e99a95b7783ff0c035c81554332ba06699","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-23 13:54:30.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-zuul-jobs-linters http://logs.openstack.org/90/672290/1/check/openstack-zuul-jobs-linters/a5b3bc6/ : SUCCESS in 8m 23s\n- project-config-grafyaml http://logs.openstack.org/90/672290/1/check/project-config-grafyaml/40e0800/ : SUCCESS in 4m 09s","accounts_in_message":[],"_revision_number":1},{"id":"6a6fd2a760e270d903f0bd3b5e95f261140505dc","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2019-07-23 13:56:32.000000000","message":"Patch Set 1:\n\n(2 comments)","accounts_in_message":[],"_revision_number":1},{"id":"a0f1e371b7eac88b5433dd9f303396939d3037a6","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2019-07-23 14:07:25.000000000","message":"Patch Set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"fd5351baef50bd885f15db0f7d41b79a504c43cc","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2019-07-23 14:07:28.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"5717ba85d02b2ad366d6804e8098037b529d2d08","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2019-07-23 14:08:14.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"30b5f4bd835a03d89534ddb9878918ceecb593eb","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-23 14:21:58.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-zuul-jobs-linters http://logs.openstack.org/90/672290/2/check/openstack-zuul-jobs-linters/9ec70e8/ : SUCCESS in 6m 40s\n- project-config-grafyaml http://logs.openstack.org/90/672290/2/check/project-config-grafyaml/300ff11/ : SUCCESS in 3m 22s","accounts_in_message":[],"_revision_number":2},{"id":"8bd7eb167fa90efbd49e7f37c8510f9808ce60aa","author":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"date":"2019-07-31 05:38:16.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"2d9c87735003119b22bbb44520c3a737a6741ef7","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-31 05:38:31.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"94109f2f1d4968cc214bc94f7c821aa88f9845d6","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-31 05:45:55.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"4c242f3249d948f13a4227645ce941372182de63","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-31 05:45:55.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-zuul-jobs-linters http://logs.openstack.org/90/672290/2/gate/openstack-zuul-jobs-linters/f376ddf/ : SUCCESS in 6m 27s\n- project-config-grafyaml http://logs.openstack.org/90/672290/2/gate/project-config-grafyaml/86ed498/ : SUCCESS in 3m 12s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"003a174f37e75f5d63030eecbbbb157ebc96149e","revisions":{"ee282bc26a11f3221d10c6fc74dc703b95f53181":{"kind":"REWORK","_number":1,"created":"2019-07-23 13:43:09.000000000","uploader":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"ref":"refs/changes/90/672290/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/90/672290/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/90/672290/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/90/672290/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/90/672290/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/90/672290/1"}}},"commit":{"parents":[{"commit":"abbace97949f6f6d178f976f338d5bb4a9235b8e","subject":"Merge \"Remove publish-openstack-manuals-developer-lang\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/abbace97949f6f6d178f976f338d5bb4a9235b8e"}]}],"author":{"name":"Slawek Kaplonski","email":"skaplons@redhat.com","date":"2019-07-23 13:42:00.000000000","tz":120},"committer":{"name":"Slawek Kaplonski","email":"skaplons@redhat.com","date":"2019-07-23 13:42:00.000000000","tz":120},"subject":"Rename \"tripleo-ci-centos-7-scenario007-standalone\" in Neutron","message":"Rename \"tripleo-ci-centos-7-scenario007-standalone\" in Neutron\n\nJob \"tripleo-ci-centos-7-scenario007-standalone\" defined in Neutron\nrepo had to be renamed to not conflict with same job proposed in\nTripleo-ci repo.\nNow this job in Neutron is called\n\"tripleo-ci-centos-7-scenario007-standalone\".\n\nChange-Id: I571afd0557f52ec9bf470c9ede46b9e3a0bb308f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/ee282bc26a11f3221d10c6fc74dc703b95f53181"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/ee282bc26a11f3221d10c6fc74dc703b95f53181"}]},"branch":"refs/heads/master"},"003a174f37e75f5d63030eecbbbb157ebc96149e":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2019-07-23 14:07:25.000000000","uploader":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"ref":"refs/changes/90/672290/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/90/672290/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/90/672290/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/90/672290/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/90/672290/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/90/672290/2"}}},"commit":{"parents":[{"commit":"abbace97949f6f6d178f976f338d5bb4a9235b8e","subject":"Merge \"Remove publish-openstack-manuals-developer-lang\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/abbace97949f6f6d178f976f338d5bb4a9235b8e"}]}],"author":{"name":"Slawek Kaplonski","email":"skaplons@redhat.com","date":"2019-07-23 13:42:00.000000000","tz":120},"committer":{"name":"Slawek Kaplonski","email":"skaplons@redhat.com","date":"2019-07-23 14:07:20.000000000","tz":0},"subject":"Rename \"tripleo-ci-centos-7-scenario007-standalone\" in Neutron","message":"Rename \"tripleo-ci-centos-7-scenario007-standalone\" in Neutron\n\nJob \"tripleo-ci-centos-7-scenario007-standalone\" defined in Neutron\nrepo had to be renamed to not conflict with same job proposed in\nTripleo-ci repo.\nNow this job in Neutron is called\n\"neutron-centos-7-tripleo-standalone\".\n\nChange-Id: I571afd0557f52ec9bf470c9ede46b9e3a0bb308f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/003a174f37e75f5d63030eecbbbb157ebc96149e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/003a174f37e75f5d63030eecbbbb157ebc96149e"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
