)]}'
{"id":"openstack%2Fproject-config~673019","triplet_id":"openstack%2Fproject-config~master~Id58f439052b4ea6b092b87682576a746433dcc27","project":"openstack/project-config","branch":"master","topic":"add_pythonjobs_for_newbranches","hashtags":[],"change_id":"Id58f439052b4ea6b092b87682576a746433dcc27","subject":"Add tooling to update python jobs on branch creation","status":"MERGED","created":"2019-07-26 16:52:22.000000000","updated":"2020-04-08 14:19:44.000000000","submitted":"2020-04-08 14:19:44.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":16,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"673019-1586355584842-a14ab837","meta_rev_id":"8e6e31bd9859a853307ebebdace3892e12baa77e","_number":673019,"virtual_id_number":673019,"owner":{"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"date":"2020-04-08 13:45:36.000000000","_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},{"value":0,"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},{"value":0,"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":0,"_account_id":8099,"name":"Graham Hayes","email":"gr@ham.ie","username":"graham"},{"value":0,"_account_id":2472,"name":"Doug Hellmann","email":"dhellmann@redhat.com","username":"doug-hellmann"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-04-08 14:19:44.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},{"value":0,"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":2,"date":"2020-04-08 13:58:52.000000000","_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},{"value":0,"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},{"value":2,"date":"2020-04-08 13:47:04.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":0,"_account_id":8099,"name":"Graham Hayes","email":"gr@ham.ie","username":"graham"},{"value":0,"_account_id":2472,"name":"Doug Hellmann","email":"dhellmann@redhat.com","username":"doug-hellmann"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},{"value":2,"date":"2020-04-08 14:11:15.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":1,"date":"2020-04-08 13:58:52.000000000","_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},{"value":0,"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},{"value":0,"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":0,"_account_id":8099,"name":"Graham Hayes","email":"gr@ham.ie","username":"graham"},{"value":0,"_account_id":2472,"name":"Doug Hellmann","email":"dhellmann@redhat.com","username":"doug-hellmann"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},{"value":0,"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},{"_account_id":2472,"name":"Doug Hellmann","email":"dhellmann@redhat.com","username":"doug-hellmann"},{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"_account_id":8099,"name":"Graham Hayes","email":"gr@ham.ie","username":"graham"},{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},{"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-07-29 13:54:07.000000000","updated_by":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"reviewer":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"state":"REVIEWER"},{"updated":"2019-09-05 14:24:05.000000000","updated_by":{"_account_id":8099,"name":"Graham Hayes","email":"gr@ham.ie","username":"graham"},"reviewer":{"_account_id":8099,"name":"Graham Hayes","email":"gr@ham.ie","username":"graham"},"state":"REVIEWER"},{"updated":"2020-04-04 17:31:31.000000000","updated_by":{"_account_id":2472,"name":"Doug Hellmann","email":"dhellmann@redhat.com","username":"doug-hellmann"},"reviewer":{"_account_id":2472,"name":"Doug Hellmann","email":"dhellmann@redhat.com","username":"doug-hellmann"},"state":"REVIEWER"},{"updated":"2020-04-08 13:45:36.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"},{"updated":"2020-04-08 13:47:04.000000000","updated_by":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"reviewer":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"state":"REVIEWER"},{"updated":"2020-04-08 13:58:52.000000000","updated_by":{"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},"reviewer":{"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},"state":"REVIEWER"},{"updated":"2020-04-08 14:11:15.000000000","updated_by":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"reviewer":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"state":"REVIEWER"},{"updated":"2020-04-08 14:19:44.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"04eccac093607327f58ba4e9075651a10fb03090","author":{"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},"date":"2019-07-26 16:52:22.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"e09906cd8c2dc65767aa25cd673889ccc6301524","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-26 17:02:17.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-zuul-jobs-linters http://logs.openstack.org/19/673019/1/check/openstack-zuul-jobs-linters/d72468d/ : FAILURE in 4m 47s","accounts_in_message":[],"_revision_number":1},{"id":"36ad8672659af7d493a28482c59b056780120db2","author":{"_account_id":2472,"name":"Doug Hellmann","email":"dhellmann@redhat.com","username":"doug-hellmann"},"date":"2019-07-26 17:04:08.000000000","message":"Patch Set 1:\n\n(3 comments)","accounts_in_message":[],"_revision_number":1},{"id":"61e32961ab9e009b25fd67fd03fee398965a8551","author":{"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},"date":"2019-07-29 07:31:04.000000000","message":"Patch Set 1:\n\n(3 comments)\n\nThis was a first draft, to catch questions. Thanks doug for the first salva of questions :)","accounts_in_message":[],"_revision_number":1},{"id":"2c1d89767012d4ecb13e8e743f47985db7001b20","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2019-07-29 12:55:16.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"90069cc5908f567ff031c700f93633e96c65077a","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2019-07-29 12:56:09.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"f55bc805809ca1c872011d910cc70029a2410c17","author":{"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},"date":"2019-07-29 13:54:00.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"4a1d42653c076d513648386e7bf55d50d9658e98","author":{"_account_id":2472,"name":"Doug Hellmann","email":"dhellmann@redhat.com","username":"doug-hellmann"},"date":"2019-07-29 13:54:27.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"c425a019cdc2bb4d19a64a8763bdbe0c920d838c","author":{"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},"date":"2019-07-29 14:20:40.000000000","message":"Patch Set 1:\n\n\u003e (1 comment)\n\nOh got it! Fine for the refactor :)","accounts_in_message":[],"_revision_number":1},{"id":"411883c20e69ed4fc0befcbe353fa352603b8b66","author":{"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},"date":"2019-09-04 14:24:35.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"b66d72c58ddae017fc238c674179f6c44a83cb23","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2019-09-04 14:33:44.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"871e9200fbd5fd34c6d23056d7df51b1ec035f6d","author":{"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},"date":"2019-09-04 14:36:18.000000000","message":"Patch Set 3: Published edit on patch set 2.","accounts_in_message":[],"_revision_number":3},{"id":"be4a5c405512769481c8e3cd8c145e23764a44e5","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2019-09-04 14:39:46.000000000","message":"Patch Set 2:\n\nThis looks good to me, but I do have concerns about the steps that need to be in place prior to creating a stable branch, and how we should handle that to avoid issues if we are not actually ready.\n\nAs it is right now, we need the following to happen before we can create a new stable branch with this in place:\n\n1) Next release named\n2) Releases repo updated with the schedule for the next series so series_status has the next series information\n3) TC approved official platforms, so that we have:\n4) Supported python runtimes defined\n5) Job definitions added for those supported runtimes for the next series name\n\nThese are not all in place yet, so if we were to merge this right now, we have branching requests that would cause issues.","accounts_in_message":[],"_revision_number":2},{"id":"8a49c25d89a80fcc0bb7a99d93fee48ae624b057","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-04 14:49:39.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/8e5c0ef1ed344d87a7993301fb62a73a : FAILURE in 5m 03s","accounts_in_message":[],"_revision_number":3},{"id":"71ffbd555381ba14f2bdaba54fca857bf844afdf","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2019-09-04 14:57:01.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"f07b5deb06e42544bb2088b9e9f512cd3d2cf144","author":{"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},"date":"2019-09-05 14:20:14.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"102160f0c079ea5e19a1a6e0ff836b3e8840c1b3","author":{"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},"date":"2019-09-20 11:38:27.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"e218fee2565630ed79310b166134fad62e50d3d1","author":{"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},"date":"2019-09-20 11:39:29.000000000","message":"Patch Set 4:\n\nnot tested.","accounts_in_message":[],"_revision_number":4},{"id":"1e198d204076be0ad0e36e081e87ae2daf303f6a","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-20 11:48:00.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/664387d61be14e68aacd015101f2f402 : FAILURE in 5m 22s","accounts_in_message":[],"_revision_number":4},{"id":"8430239f71a18d6ff237ae2e647cf29a2369031b","author":{"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},"date":"2019-09-20 14:09:27.000000000","message":"Patch Set 4: Workflow-1","accounts_in_message":[],"_revision_number":4},{"id":"5291b05a430c59536bd11d28c15d82680fe859e9","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2019-09-22 09:11:58.000000000","message":"Patch Set 4: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"efa533a9a3cc6f4e2f66b955e1099e09a9f3befc","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2019-09-22 09:12:54.000000000","message":"Patch Set 5: Published edit on patch set 4.","accounts_in_message":[],"_revision_number":5},{"id":"9d528da328031793fed5d5ddd006d9700d156296","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2019-09-22 09:17:03.000000000","message":"Patch Set 5:\n\nBefore we merge this, openstack-zuul-jobs needs updating with the new ussuri jobs:\n\nRight now we only have:\n\nzuul.d/project-templates.yaml:    name: openstack-python3-train-jobs\nzuul.d/project-templates.yaml:    name: openstack-python3-train-jobs-horizon\nzuul.d/project-templates.yaml:    name: openstack-python3-train-jobs-neutron\nzuul.d/project-templates.yaml:    name: openstack-python3-train-jobs-ceilometer\n\nPlease provide patch and add a depends-on for it.","accounts_in_message":[],"_revision_number":5},{"id":"123aefe7a4fdb33686128270851e98ea22d609db","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-22 09:17:03.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/d4b222f2214f4a709a184ff99f0c8fe5 : FAILURE in 3m 43s","accounts_in_message":[],"_revision_number":5},{"id":"14d6e1e71e51ee1934b44caeb91b6a5974665eee","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2019-09-22 09:18:04.000000000","message":"Patch Set 5: Code-Review-1\n\nAnd please make pep8 happy ;)","accounts_in_message":[],"_revision_number":5},{"id":"2898ce261e311c1cbdfbe31adeb44553c701d56e","author":{"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},"date":"2019-09-23 07:56:46.000000000","message":"Patch Set 5: Workflow-1\n\nYeah, sorry it\u0027s still WIP when I have extra time.\nI know I have to fix bashate and others, but we had other things to fix first.","accounts_in_message":[],"_revision_number":5},{"id":"ce050f880f1eacceba84d66205d6b8d0b8e6900a","author":{"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},"date":"2020-01-07 12:10:36.000000000","message":"Patch Set 5:\n\nThis will need a refresh for the future.","accounts_in_message":[],"_revision_number":5},{"id":"74d8ae93eab7faf071fecbbc588e4ec837c1df31","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-04-04 16:43:44.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"dba5f6bb38331640522d15f89bcf22c45a4f188e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-04 17:04:01.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/d5eaa4a85b49412c967e71706f4d8ae3 : SUCCESS in 7m 52s","accounts_in_message":[],"_revision_number":6},{"id":"5173f585af55a6950f6d59fab002b9fa72f1f171","author":{"_account_id":2472,"name":"Doug Hellmann","email":"dhellmann@redhat.com","username":"doug-hellmann"},"date":"2020-04-04 17:31:31.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"d1b2623ffc6d80c7709d397df646d9b79bc91711","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-04-08 13:37:51.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"94621ee7644659a0ea5ec7431a247fbe501cf312","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-04-08 13:44:23.000000000","message":"Uploaded patch set 7.","accounts_in_message":[],"_revision_number":7},{"id":"865e4941ccc936fc76d09f45963124d97153b5e6","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-04-08 13:45:36.000000000","message":"Patch Set 7:\n\nClarified changed file check. I have also tested this locally (with git review line commented out) and works as expected.\n\nWould be good to get this in as we enter the transition phase from ussuri to victoria.","accounts_in_message":[],"_revision_number":7},{"id":"34f2216ba86c06184082a13db2477c043c90f3af","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2020-04-08 13:47:04.000000000","message":"Patch Set 7: Code-Review+2","accounts_in_message":[],"_revision_number":7},{"id":"71ae348c056f4a406ef9039adc25f5869d13c447","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-08 13:54:42.000000000","message":"Patch Set 7: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/b0bb091eaf054f07ab5c7b91fe232ead : SUCCESS in 8m 33s","accounts_in_message":[],"_revision_number":7},{"id":"9f7ef1b8b23e50122ce3ae772bb7e2c19924b636","author":{"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},"date":"2020-04-08 13:58:52.000000000","message":"Patch Set 7: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":7},{"id":"26ae19664b8abc74bac03bc10c54f14157e7f550","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-08 14:03:31.000000000","message":"Patch Set 7: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":7},{"id":"b9cbf6c0abaf25f02290b2b3eb3aac4f0bdf8bdc","author":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"date":"2020-04-08 14:11:15.000000000","message":"Patch Set 7: Code-Review+2","accounts_in_message":[],"_revision_number":7},{"id":"8841447c7832afae8356960d167c7236228c95de","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-08 14:19:44.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":7},{"id":"a2b2620dc881ab9c6ec13814650f22f83e8a3c93","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-08 14:19:44.000000000","message":"Patch Set 7: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/a00512959b8e4063a1e51f89bb951ab3 : SUCCESS in 10m 02s","accounts_in_message":[],"_revision_number":7}],"current_revision_number":7,"current_revision":"6938731988745a1352116654d43fc16125d81b22","revisions":{"4d30ecf00c3d5637a8d3feec2a6dec3dafe7ed40":{"kind":"REWORK","_number":1,"created":"2019-07-26 16:52:22.000000000","uploader":{"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},"ref":"refs/changes/19/673019/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/19/673019/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/19/673019/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/19/673019/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/19/673019/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/19/673019/1"}}},"commit":{"parents":[{"commit":"a258948508a96817943575181fab65a8171b01e4","subject":"Merge \"Cleanup in-tree removed jobs\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/a258948508a96817943575181fab65a8171b01e4"}]}],"author":{"name":"Jean-Philippe Evrard","email":"jean-philippe@evrard.me","date":"2019-07-26 16:49:34.000000000","tz":120},"committer":{"name":"Jean-Philippe Evrard","email":"jean-philippe@evrard.me","date":"2019-07-26 16:52:20.000000000","tz":120},"subject":"[WIP] Add tooling to update python jobs on branch creation","message":"[WIP] Add tooling to update python jobs on branch creation\n\nCurrently, we need to update jobs manually after a branch is\ncreated.\n\nWhen a project is branched, the master branch should then\nbe pointing to the new named python3 tests.\n\nThis should do it.\n\nChange-Id: Id58f439052b4ea6b092b87682576a746433dcc27\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/4d30ecf00c3d5637a8d3feec2a6dec3dafe7ed40"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/4d30ecf00c3d5637a8d3feec2a6dec3dafe7ed40"}]},"branch":"refs/heads/master"},"8d4ec17d063c012187de1d286963f66ec0a5cd65":{"kind":"REWORK","_number":2,"created":"2019-09-04 14:24:35.000000000","uploader":{"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},"ref":"refs/changes/19/673019/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/19/673019/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/19/673019/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/19/673019/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/19/673019/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/19/673019/2"}}},"commit":{"parents":[{"commit":"a258948508a96817943575181fab65a8171b01e4","subject":"Merge \"Cleanup in-tree removed jobs\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/a258948508a96817943575181fab65a8171b01e4"}]}],"author":{"name":"Jean-Philippe Evrard","email":"jean-philippe@evrard.me","date":"2019-07-26 16:49:34.000000000","tz":120},"committer":{"name":"Jean-Philippe Evrard","email":"jean-philippe@evrard.me","date":"2019-09-04 14:24:13.000000000","tz":120},"subject":"Add tooling to update python jobs on branch creation","message":"Add tooling to update python jobs on branch creation\n\nCurrently, we need to update jobs manually after a branch is\ncreated.\n\nWhen a project is branched, the master branch should then\nbe pointing to the new named python3 tests.\n\nThis should do it.\n\nChange-Id: Id58f439052b4ea6b092b87682576a746433dcc27\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/8d4ec17d063c012187de1d286963f66ec0a5cd65"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/8d4ec17d063c012187de1d286963f66ec0a5cd65"}]},"branch":"refs/heads/master"},"c8ba99dbef06d5e65a143cc2b9944ae20e6983a7":{"kind":"REWORK","_number":3,"created":"2019-09-04 14:36:18.000000000","uploader":{"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},"ref":"refs/changes/19/673019/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/19/673019/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/19/673019/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/19/673019/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/19/673019/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/19/673019/3"}}},"commit":{"parents":[{"commit":"a258948508a96817943575181fab65a8171b01e4","subject":"Merge \"Cleanup in-tree removed jobs\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/a258948508a96817943575181fab65a8171b01e4"}]}],"author":{"name":"Jean-Philippe Evrard","email":"jean-philippe@evrard.me","date":"2019-07-26 16:49:34.000000000","tz":120},"committer":{"name":"Jean-Philippe Evrard","email":"jean-philippe@evrard.me","date":"2019-09-04 14:36:15.000000000","tz":0},"subject":"Add tooling to update python jobs on branch creation","message":"Add tooling to update python jobs on branch creation\n\nCurrently, we need to update jobs manually after a branch is\ncreated.\n\nWhen a project is branched, the master branch should then\nbe pointing to the new named python3 tests.\n\nThis should do it.\n\nChange-Id: Id58f439052b4ea6b092b87682576a746433dcc27\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/c8ba99dbef06d5e65a143cc2b9944ae20e6983a7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/c8ba99dbef06d5e65a143cc2b9944ae20e6983a7"}]},"branch":"refs/heads/master"},"98e0a3d2ffe464942116f23ff0d8523725131b2d":{"kind":"REWORK","_number":4,"created":"2019-09-20 11:38:27.000000000","uploader":{"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},"ref":"refs/changes/19/673019/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/19/673019/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/19/673019/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/19/673019/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/19/673019/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/19/673019/4"}}},"commit":{"parents":[{"commit":"a258948508a96817943575181fab65a8171b01e4","subject":"Merge \"Cleanup in-tree removed jobs\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/a258948508a96817943575181fab65a8171b01e4"}]}],"author":{"name":"Jean-Philippe Evrard","email":"jean-philippe@evrard.me","date":"2019-07-26 16:49:34.000000000","tz":120},"committer":{"name":"Jean-Philippe Evrard","email":"jean-philippe@evrard.me","date":"2019-09-20 11:38:00.000000000","tz":120},"subject":"Add tooling to update python jobs on branch creation","message":"Add tooling to update python jobs on branch creation\n\nCurrently, we need to update jobs manually after a branch is\ncreated.\n\nWhen a project is branched, the master branch should then\nbe pointing to the new named python3 tests.\n\nThis should do it.\n\nChange-Id: Id58f439052b4ea6b092b87682576a746433dcc27\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/98e0a3d2ffe464942116f23ff0d8523725131b2d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/98e0a3d2ffe464942116f23ff0d8523725131b2d"}]},"branch":"refs/heads/master"},"c15fc858f6457e647b61d069a0f712b6fe811dda":{"kind":"REWORK","_number":5,"created":"2019-09-22 09:12:54.000000000","uploader":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"ref":"refs/changes/19/673019/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/19/673019/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/19/673019/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/19/673019/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/19/673019/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/19/673019/5"}}},"commit":{"parents":[{"commit":"a258948508a96817943575181fab65a8171b01e4","subject":"Merge \"Cleanup in-tree removed jobs\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/a258948508a96817943575181fab65a8171b01e4"}]}],"author":{"name":"Jean-Philippe Evrard","email":"jean-philippe@evrard.me","date":"2019-07-26 16:49:34.000000000","tz":120},"committer":{"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","date":"2019-09-22 09:12:43.000000000","tz":0},"subject":"Add tooling to update python jobs on branch creation","message":"Add tooling to update python jobs on branch creation\n\nCurrently, we need to update jobs manually after a branch is\ncreated.\n\nWhen a project is branched, the master branch should then\nbe pointing to the new named python3 tests.\n\nThis should do it.\n\nChange-Id: Id58f439052b4ea6b092b87682576a746433dcc27\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/c15fc858f6457e647b61d069a0f712b6fe811dda"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/c15fc858f6457e647b61d069a0f712b6fe811dda"}]},"branch":"refs/heads/master"},"2c58db384a2328ca0bc9b18bc6c7c715a44a7883":{"kind":"REWORK","_number":6,"created":"2020-04-04 16:43:44.000000000","uploader":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"ref":"refs/changes/19/673019/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/19/673019/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/19/673019/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/19/673019/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/19/673019/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/19/673019/6"}}},"commit":{"parents":[{"commit":"a8a41f162f469489f670a552cf3d42b521eceae8","subject":"Install gpg tooling on dib images","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/a8a41f162f469489f670a552cf3d42b521eceae8"}]}],"author":{"name":"Jean-Philippe Evrard","email":"jean-philippe@evrard.me","date":"2019-07-26 16:49:34.000000000","tz":120},"committer":{"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","date":"2020-04-04 16:43:31.000000000","tz":-300},"subject":"Add tooling to update python jobs on branch creation","message":"Add tooling to update python jobs on branch creation\n\nCurrently, we need to update jobs manually after a branch is\ncreated.\n\nWhen a project is branched, the master branch should then\nbe pointing to the new named python3 tests.\n\nThis should do it.\n\nChange-Id: Id58f439052b4ea6b092b87682576a746433dcc27\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/2c58db384a2328ca0bc9b18bc6c7c715a44a7883"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/2c58db384a2328ca0bc9b18bc6c7c715a44a7883"}]},"branch":"refs/heads/master"},"6938731988745a1352116654d43fc16125d81b22":{"kind":"REWORK","_number":7,"created":"2020-04-08 13:44:23.000000000","uploader":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"ref":"refs/changes/19/673019/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/19/673019/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/19/673019/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/19/673019/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/19/673019/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/19/673019/7"}}},"commit":{"parents":[{"commit":"a8a41f162f469489f670a552cf3d42b521eceae8","subject":"Install gpg tooling on dib images","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/a8a41f162f469489f670a552cf3d42b521eceae8"}]}],"author":{"name":"Jean-Philippe Evrard","email":"jean-philippe@evrard.me","date":"2019-07-26 16:49:34.000000000","tz":120},"committer":{"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","date":"2020-04-08 13:44:14.000000000","tz":-300},"subject":"Add tooling to update python jobs on branch creation","message":"Add tooling to update python jobs on branch creation\n\nCurrently, we need to update jobs manually after a branch is\ncreated.\n\nWhen a project is branched, the master branch should then\nbe pointing to the new named python3 tests.\n\nThis should do it.\n\nChange-Id: Id58f439052b4ea6b092b87682576a746433dcc27\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/6938731988745a1352116654d43fc16125d81b22"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/6938731988745a1352116654d43fc16125d81b22"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
