)]}'
{"id":"openstack%2Fproject-config~673304","triplet_id":"openstack%2Fproject-config~master~Ifb440750c18aef96b42080cc50799e5ed8703619","project":"openstack/project-config","branch":"master","hashtags":[],"change_id":"Ifb440750c18aef96b42080cc50799e5ed8703619","subject":"Default zuul tenant to use ansible 2.8 for jobs","status":"MERGED","created":"2019-07-29 14:40:18.000000000","updated":"2019-08-06 15:47:43.000000000","submitted":"2019-08-06 15:47:43.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"673304-1565106463638-9b622d95","meta_rev_id":"1733395d06766a478ed6d6acf092b855839e95c8","_number":673304,"virtual_id_number":673304,"owner":{"_account_id":4162,"name":"Paul Belanger","email":"pabelanger@redhat.com","username":"pabelanger"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":1,"name":"James E. Blair","email":"jim@acmegating.com","username":"corvus"},{"value":0,"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"value":0,"_account_id":16068,"name":"Tobias Henkel","email":"tobias.henkel@bmw.de","username":"tobias.henkel"},{"value":2,"date":"2019-08-06 15:47:40.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":1,"name":"James E. Blair","email":"jim@acmegating.com","username":"corvus"},"all":[{"value":2,"date":"2019-07-29 17:48:43.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":1,"name":"James E. Blair","email":"jim@acmegating.com","username":"corvus"},{"value":2,"date":"2019-08-06 15:33:24.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"value":1,"date":"2019-08-06 14:12:42.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":16068,"name":"Tobias Henkel","email":"tobias.henkel@bmw.de","username":"tobias.henkel"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"all":[{"value":0,"_account_id":1,"name":"James E. Blair","email":"jim@acmegating.com","username":"corvus"},{"value":1,"date":"2019-08-06 15:33:24.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"value":0,"_account_id":16068,"name":"Tobias Henkel","email":"tobias.henkel@bmw.de","username":"tobias.henkel"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1,"name":"James E. Blair","email":"jim@acmegating.com","username":"corvus"},{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"_account_id":16068,"name":"Tobias Henkel","email":"tobias.henkel@bmw.de","username":"tobias.henkel"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-07-29 17:48:43.000000000","updated_by":{"_account_id":1,"name":"James E. Blair","email":"jim@acmegating.com","username":"corvus"},"reviewer":{"_account_id":1,"name":"James E. Blair","email":"jim@acmegating.com","username":"corvus"},"state":"REVIEWER"},{"updated":"2019-08-06 14:12:42.000000000","updated_by":{"_account_id":16068,"name":"Tobias Henkel","email":"tobias.henkel@bmw.de","username":"tobias.henkel"},"reviewer":{"_account_id":16068,"name":"Tobias Henkel","email":"tobias.henkel@bmw.de","username":"tobias.henkel"},"state":"REVIEWER"},{"updated":"2019-08-06 15:33:24.000000000","updated_by":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"reviewer":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"state":"REVIEWER"},{"updated":"2019-08-06 15:47:40.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"365daa59809c486257857b89dd20c574a4a78051","author":{"_account_id":4162,"name":"Paul Belanger","email":"pabelanger@redhat.com","username":"pabelanger"},"date":"2019-07-29 14:40:18.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"99f75b7c544d5d982b444de3413cabf93afc8452","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-29 14:54:33.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-zuul-jobs-linters http://logs.openstack.org/04/673304/1/check/openstack-zuul-jobs-linters/eace69d/ : SUCCESS in 9m 14s\n- project-config-gerrit http://logs.openstack.org/04/673304/1/check/project-config-gerrit/f54e5d9/ : SUCCESS in 5m 05s","accounts_in_message":[],"_revision_number":1},{"id":"6f019bb4c94019d74563e0e78cdbfa0740ce8f7a","author":{"_account_id":1,"name":"James E. Blair","email":"jim@acmegating.com","username":"corvus"},"date":"2019-07-29 17:48:43.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"c92952bc98176e8d4364fd00fcfa6edc1d88edc0","author":{"_account_id":16068,"name":"Tobias Henkel","email":"tobias.henkel@bmw.de","username":"tobias.henkel"},"date":"2019-08-06 14:12:42.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"cd0283eea6207072299fe1706b0f77ebfacf6efe","author":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"date":"2019-08-06 15:33:24.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"338c35e507f9230bf945efd6d6d05d14b283204c","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-06 15:33:42.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"2f39793f49f6e05fe48ec5028c1b179079311fa1","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-06 15:47:40.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-zuul-jobs-linters https://logs.opendev.org/04/673304/1/gate/openstack-zuul-jobs-linters/a738a58/ : SUCCESS in 7m 49s\n- project-config-gerrit https://logs.opendev.org/04/673304/1/gate/project-config-gerrit/6aacd17/ : SUCCESS in 4m 10s","accounts_in_message":[],"_revision_number":1},{"id":"1733395d06766a478ed6d6acf092b855839e95c8","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-06 15:47:43.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"9fdebd38e34cab742a718c775b5774ce37304131","revisions":{"9fdebd38e34cab742a718c775b5774ce37304131":{"kind":"REWORK","_number":1,"created":"2019-07-29 14:40:18.000000000","uploader":{"_account_id":4162,"name":"Paul Belanger","email":"pabelanger@redhat.com","username":"pabelanger"},"ref":"refs/changes/04/673304/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/04/673304/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/04/673304/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/04/673304/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/04/673304/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/04/673304/1"}}},"commit":{"parents":[{"commit":"e02cdfa1f489838aa2bb4b449467c0ac4d83645f","subject":"Merge \"Remove puppet-crane\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/e02cdfa1f489838aa2bb4b449467c0ac4d83645f"}]}],"author":{"name":"Paul Belanger","email":"pabelanger@redhat.com","date":"2019-07-29 14:38:18.000000000","tz":-240},"committer":{"name":"Paul Belanger","email":"pabelanger@redhat.com","date":"2019-07-29 14:38:18.000000000","tz":-240},"subject":"Default zuul tenant to use ansible 2.8 for jobs","message":"Default zuul tenant to use ansible 2.8 for jobs\n\nAfter some basic testing of jobs in the zuul tenant, we don\u0027t appear to\nhave any issues using ansible 2.8. With that in mind, switch the default\nfor all jobs to ansible 2.8. We also only do this on the zuul tenant, to\nensure this tenant works properly before moving to other tenants.\n\nChange-Id: Ifb440750c18aef96b42080cc50799e5ed8703619\nSigned-off-by: Paul Belanger \u003cpabelanger@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/9fdebd38e34cab742a718c775b5774ce37304131"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/9fdebd38e34cab742a718c775b5774ce37304131"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
