)]}'
{"id":"openstack%2Fproject-config~673898","triplet_id":"openstack%2Fproject-config~master~I40c4c40292bbb9543cbcbd560731d89610f9a99d","project":"openstack/project-config","branch":"master","topic":"new-project","hashtags":[],"change_id":"I40c4c40292bbb9543cbcbd560731d89610f9a99d","subject":"New project request: airship/porthole","status":"MERGED","created":"2019-07-31 19:37:53.000000000","updated":"2019-08-05 16:48:24.000000000","submitted":"2019-08-05 16:48:24.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"673898-1565023704519-3342d8ab","meta_rev_id":"3aa0eec5a783364e9226ace01d6872403c4b6141","_number":673898,"virtual_id_number":673898,"owner":{"_account_id":22477,"name":"Matt McEuen","email":"matt.mceuen@att.com","username":"mattmceuen"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"value":0,"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":0,"_account_id":29144,"name":"PRATEEK REDDY DODDA","email":"prateek.reddy.dodda@att.com","username":"pd2839"},{"value":0,"_account_id":29161,"name":"Rahul Khiyani","email":"rahul.khiyani@att.com","username":"rk0850"},{"value":0,"_account_id":29131,"name":"diwakar thyagaraj","email":"diwakar.chitoor.thyagaraj@att.com","username":"dt241s@att.com"},{"value":2,"date":"2019-08-05 16:48:24.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"all":[{"value":2,"date":"2019-08-05 16:38:08.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"value":2,"date":"2019-07-31 19:57:18.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":1,"date":"2019-08-01 17:47:19.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":29144,"name":"PRATEEK REDDY DODDA","email":"prateek.reddy.dodda@att.com","username":"pd2839"},{"value":1,"date":"2019-08-01 18:06:44.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":29161,"name":"Rahul Khiyani","email":"rahul.khiyani@att.com","username":"rk0850"},{"value":1,"date":"2019-08-01 17:34:28.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":29131,"name":"diwakar thyagaraj","email":"diwakar.chitoor.thyagaraj@att.com","username":"dt241s@att.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"all":[{"value":1,"date":"2019-08-05 16:38:08.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"value":0,"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":0,"_account_id":29144,"name":"PRATEEK REDDY DODDA","email":"prateek.reddy.dodda@att.com","username":"pd2839"},{"value":0,"_account_id":29161,"name":"Rahul Khiyani","email":"rahul.khiyani@att.com","username":"rk0850"},{"value":0,"_account_id":29131,"name":"diwakar thyagaraj","email":"diwakar.chitoor.thyagaraj@att.com","username":"dt241s@att.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":29131,"name":"diwakar thyagaraj","email":"diwakar.chitoor.thyagaraj@att.com","username":"dt241s@att.com"},{"_account_id":29144,"name":"PRATEEK REDDY DODDA","email":"prateek.reddy.dodda@att.com","username":"pd2839"},{"_account_id":29161,"name":"Rahul Khiyani","email":"rahul.khiyani@att.com","username":"rk0850"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-07-31 19:57:18.000000000","updated_by":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"reviewer":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"state":"REVIEWER"},{"updated":"2019-08-01 17:34:28.000000000","updated_by":{"_account_id":29131,"name":"diwakar thyagaraj","email":"diwakar.chitoor.thyagaraj@att.com","username":"dt241s@att.com"},"reviewer":{"_account_id":29131,"name":"diwakar thyagaraj","email":"diwakar.chitoor.thyagaraj@att.com","username":"dt241s@att.com"},"state":"REVIEWER"},{"updated":"2019-08-01 17:47:19.000000000","updated_by":{"_account_id":29144,"name":"PRATEEK REDDY DODDA","email":"prateek.reddy.dodda@att.com","username":"pd2839"},"reviewer":{"_account_id":29144,"name":"PRATEEK REDDY DODDA","email":"prateek.reddy.dodda@att.com","username":"pd2839"},"state":"REVIEWER"},{"updated":"2019-08-01 18:06:44.000000000","updated_by":{"_account_id":29161,"name":"Rahul Khiyani","email":"rahul.khiyani@att.com","username":"rk0850"},"reviewer":{"_account_id":29161,"name":"Rahul Khiyani","email":"rahul.khiyani@att.com","username":"rk0850"},"state":"REVIEWER"},{"updated":"2019-08-05 16:38:08.000000000","updated_by":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"reviewer":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"state":"REVIEWER"},{"updated":"2019-08-05 16:48:24.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"d03bbf80e62d4ed1c30afc4174c488761b6ef63b","author":{"_account_id":22477,"name":"Matt McEuen","email":"matt.mceuen@att.com","username":"mattmceuen"},"date":"2019-07-31 19:37:53.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"df89d073017c1246579f0978a0905d5a652ec966","author":{"_account_id":22477,"name":"Matt McEuen","email":"matt.mceuen@att.com","username":"mattmceuen"},"date":"2019-07-31 19:38:19.000000000","message":"Topic changed from new-project-porthole to new-project","accounts_in_message":[],"_revision_number":1},{"id":"076be43a74d798bbc34c52713d662c537dbe411b","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-31 19:51:04.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-zuul-jobs-linters http://logs.openstack.org/98/673898/1/check/openstack-zuul-jobs-linters/980d39a/ : SUCCESS in 6m 27s\n- project-config-gerrit http://logs.openstack.org/98/673898/1/check/project-config-gerrit/c712edd/ : SUCCESS in 4m 05s","accounts_in_message":[],"_revision_number":1},{"id":"19df81da64dfc574d726a2b0fd9987bde3a725fb","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2019-07-31 19:57:18.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"8cf037bae9338a649073b5ee83a8766a5e4870ff","author":{"_account_id":29131,"name":"diwakar thyagaraj","email":"diwakar.chitoor.thyagaraj@att.com","username":"dt241s@att.com"},"date":"2019-08-01 17:34:28.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"9271aef0a8a3cddd3c6b8dd6695f4aa7c732ee47","author":{"_account_id":29144,"name":"PRATEEK REDDY DODDA","email":"prateek.reddy.dodda@att.com","username":"pd2839"},"date":"2019-08-01 17:47:19.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"b3d62a36dfc365d00171e339276f3116f033a957","author":{"_account_id":29161,"name":"Rahul Khiyani","email":"rahul.khiyani@att.com","username":"rk0850"},"date":"2019-08-01 18:06:44.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"3dd405121400bdb9b79611d26596beed767baabc","author":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"date":"2019-08-05 16:38:08.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"ab8ac16697b95fbb73317239867b14b61a67f019","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-05 16:38:22.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"6e04249618fc6e3b199cbdf4d09e9d02b781d3eb","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-05 16:48:24.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"3aa0eec5a783364e9226ace01d6872403c4b6141","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-05 16:48:24.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-zuul-jobs-linters https://logs.opendev.org/98/673898/1/gate/openstack-zuul-jobs-linters/d2b8448/ : SUCCESS in 8m 23s\n- project-config-gerrit https://logs.opendev.org/98/673898/1/gate/project-config-gerrit/b91329f/ : SUCCESS in 3m 57s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"d3590fcc5921555fa9e6858649bbad7ad661be1f","revisions":{"d3590fcc5921555fa9e6858649bbad7ad661be1f":{"kind":"REWORK","_number":1,"created":"2019-07-31 19:37:53.000000000","uploader":{"_account_id":22477,"name":"Matt McEuen","email":"matt.mceuen@att.com","username":"mattmceuen"},"ref":"refs/changes/98/673898/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/98/673898/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/98/673898/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/98/673898/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/98/673898/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/98/673898/1"}}},"commit":{"parents":[{"commit":"8150b18ea43070a0861c66a072f95980f2f0e430","subject":"Merge \"Add new job \"tempest-ipv6-only\" to neutron CI dashboard\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/8150b18ea43070a0861c66a072f95980f2f0e430"}]}],"author":{"name":"Matt McEuen","email":"matt.mceuen@att.com","date":"2019-07-31 19:35:45.000000000","tz":-300},"committer":{"name":"Matt McEuen","email":"matt.mceuen@att.com","date":"2019-07-31 19:35:45.000000000","tz":-300},"subject":"New project request: airship/porthole","message":"New project request: airship/porthole\n\nThis moves the Porthole project, which has been incubating in\ngithub, into the Airship namespace.  Porthole is a project that\nprovides a collection of utility containers which can be used\nto access various Airship and OpenStack-Helm component CLIs\n(e.g. etcdctl, psql...), and can be configured for enterprise\nuse cases that require fine-grained RBAC controls.\n\nChange-Id: I40c4c40292bbb9543cbcbd560731d89610f9a99d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/d3590fcc5921555fa9e6858649bbad7ad661be1f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/d3590fcc5921555fa9e6858649bbad7ad661be1f"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
