)]}'
{"id":"openstack%2Fproject-config~675255","triplet_id":"openstack%2Fproject-config~master~I982f20385889bb1d9721b18cca179e49e39b903a","project":"openstack/project-config","branch":"master","topic":"linting","hashtags":[],"change_id":"I982f20385889bb1d9721b18cca179e49e39b903a","subject":"Bump ansible-lint to version 4","status":"ABANDONED","created":"2019-08-08 07:04:18.000000000","updated":"2019-08-27 15:59:54.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"18788c08d8dc8790aef132a59e983d98e5db0541","_number":675255,"virtual_id_number":675255,"owner":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":1,"date":"2019-08-09 04:30:24.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2019-08-09 05:27:49.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-08-09 04:30:24.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-08-09 05:27:49.000000000","updated_by":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"reviewer":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"state":"REVIEWER"}],"messages":[{"id":"06f0f45b886c6c759f5950bd220f12a556185e1e","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2019-08-08 07:04:18.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"5d37f5bbe9447afa2a5a5a38de8c78af99c1062c","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2019-08-08 07:10:00.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"e944e71b07958e3d9d4f94da788ba91c7b86b53e","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-08 07:23:27.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- build-openstack-specs-site https://logs.opendev.org/55/675255/2/check/build-openstack-specs-site/4034f28/ : SUCCESS in 3m 58s\n- openstack-zuul-jobs-linters https://logs.opendev.org/55/675255/2/check/openstack-zuul-jobs-linters/a89674c/ : FAILURE in 7m 58s\n- project-config-dib https://logs.opendev.org/55/675255/2/check/project-config-dib/e2da45a/ : SUCCESS in 3m 12s\n- project-config-gerrit https://logs.opendev.org/55/675255/2/check/project-config-gerrit/2a443d0/ : SUCCESS in 3m 58s\n- project-config-grafyaml https://logs.opendev.org/55/675255/2/check/project-config-grafyaml/0672e10/ : SUCCESS in 4m 42s\n- project-config-infra-docs-index https://logs.opendev.org/55/675255/2/check/project-config-infra-docs-index/fbc42dc/ : SUCCESS in 3m 04s\n- project-config-irc-access https://logs.opendev.org/55/675255/2/check/project-config-irc-access/7d818c7/ : SUCCESS in 7m 14s\n- project-config-nodepool https://logs.opendev.org/55/675255/2/check/project-config-nodepool/777274d/ : SUCCESS in 3m 52s","accounts_in_message":[],"_revision_number":2},{"id":"61c5fbda5c10a4375156cc87072340c880a73bab","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2019-08-08 07:43:56.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"7c580afd39083aeb1729522e4f13e0c9f9bda462","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-08 08:01:34.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- build-openstack-specs-site https://logs.opendev.org/55/675255/3/check/build-openstack-specs-site/81f98cb/ : SUCCESS in 5m 01s\n- openstack-zuul-jobs-linters https://logs.opendev.org/55/675255/3/check/openstack-zuul-jobs-linters/2ca7b0f/ : FAILURE in 8m 00s\n- project-config-dib https://logs.opendev.org/55/675255/3/check/project-config-dib/88f63f5/ : SUCCESS in 3m 38s\n- project-config-gerrit https://logs.opendev.org/55/675255/3/check/project-config-gerrit/aba74cd/ : SUCCESS in 4m 09s\n- project-config-grafyaml https://logs.opendev.org/55/675255/3/check/project-config-grafyaml/5587bc8/ : SUCCESS in 4m 08s\n- project-config-infra-docs-index https://logs.opendev.org/55/675255/3/check/project-config-infra-docs-index/edc65f7/ : SUCCESS in 4m 04s\n- project-config-irc-access https://logs.opendev.org/55/675255/3/check/project-config-irc-access/61310be/ : SUCCESS in 7m 00s\n- project-config-nodepool https://logs.opendev.org/55/675255/3/check/project-config-nodepool/3944471/ : SUCCESS in 4m 31s","accounts_in_message":[],"_revision_number":3},{"id":"7a463f3454620ad36b8b47d730441e1b6b236cd8","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2019-08-08 08:05:26.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"8826f06e99797e40efa7701f1abcb1909d30b6d6","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2019-08-08 08:06:32.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"a1a9df5a40aa09c6b716b4707acc697651b7e570","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-08 08:19:58.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\n\n- build-openstack-specs-site https://logs.opendev.org/55/675255/5/check/build-openstack-specs-site/3fd7bd3/ : SUCCESS in 3m 05s\n- openstack-zuul-jobs-linters https://logs.opendev.org/55/675255/5/check/openstack-zuul-jobs-linters/117be77/ : SUCCESS in 8m 10s\n- project-config-dib https://logs.opendev.org/55/675255/5/check/project-config-dib/2f9a14c/ : SUCCESS in 4m 28s\n- project-config-gerrit https://logs.opendev.org/55/675255/5/check/project-config-gerrit/752e658/ : SUCCESS in 3m 54s\n- project-config-grafyaml https://logs.opendev.org/55/675255/5/check/project-config-grafyaml/113ef1d/ : SUCCESS in 3m 24s\n- project-config-infra-docs-index https://logs.opendev.org/55/675255/5/check/project-config-infra-docs-index/e05e0a5/ : SUCCESS in 3m 21s\n- project-config-irc-access https://logs.opendev.org/55/675255/5/check/project-config-irc-access/04d4c50/ : SUCCESS in 7m 37s\n- project-config-nodepool https://logs.opendev.org/55/675255/5/check/project-config-nodepool/01743f5/ : SUCCESS in 5m 13s","accounts_in_message":[],"_revision_number":5},{"id":"4790b916570f81d33736b5b871e78ae437164840","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2019-08-09 04:20:50.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"6603196694b2275a7f760992d43c29add151cc70","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2019-08-09 04:21:54.000000000","message":"Uploaded patch set 7.","accounts_in_message":[],"_revision_number":7},{"id":"c565126ca752aeaf40ec163f6cf7ce450f9c486e","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-09 04:30:24.000000000","message":"Patch Set 7: Verified+1\n\nBuild succeeded (check pipeline).\n\n- build-openstack-specs-site https://logs.opendev.org/55/675255/7/check/build-openstack-specs-site/6726a5b/ : SUCCESS in 3m 31s\n- openstack-zuul-jobs-linters https://logs.opendev.org/55/675255/7/check/openstack-zuul-jobs-linters/bbd7a27/ : SUCCESS in 7m 27s\n- project-config-dib https://logs.opendev.org/55/675255/7/check/project-config-dib/e17df89/ : SUCCESS in 2m 53s\n- project-config-gerrit https://logs.opendev.org/55/675255/7/check/project-config-gerrit/3f43e84/ : SUCCESS in 3m 40s\n- project-config-grafyaml https://logs.opendev.org/55/675255/7/check/project-config-grafyaml/eac8a8b/ : SUCCESS in 2m 36s\n- project-config-infra-docs-index https://logs.opendev.org/55/675255/7/check/project-config-infra-docs-index/189de9c/ : SUCCESS in 2m 56s\n- project-config-irc-access https://logs.opendev.org/55/675255/7/check/project-config-irc-access/0f91ec7/ : SUCCESS in 6m 46s\n- project-config-nodepool https://logs.opendev.org/55/675255/7/check/project-config-nodepool/aee3bab/ : SUCCESS in 4m 13s","accounts_in_message":[],"_revision_number":7},{"id":"9ec528f9925ce62d11cad8cb0aa4a371d38f9a73","author":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"date":"2019-08-09 05:27:49.000000000","message":"Patch Set 7: Code-Review+2","accounts_in_message":[],"_revision_number":7},{"id":"11d4529f347df535f1440ddece536ad4685c1ad6","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2019-08-27 15:59:54.000000000","message":"Abandoned\n\nno interest for now to take this","accounts_in_message":[],"_revision_number":7}],"current_revision_number":7,"current_revision":"d1d284175df0ace5ee3e5be56684f861364231e1","revisions":{"58fb09ff39c63f83970a80b26877a862ee0723e4":{"kind":"REWORK","_number":1,"created":"2019-08-08 07:04:18.000000000","uploader":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"ref":"refs/changes/55/675255/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/55/675255/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/55/675255/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/55/675255/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/55/675255/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/55/675255/1"}}},"commit":{"parents":[{"commit":"4972c5fc8f9209140a5ef16b581deaa44b426af2","subject":"Merge \"projects: add vexxhost/ansible-role-openmanage\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/4972c5fc8f9209140a5ef16b581deaa44b426af2"}]}],"author":{"name":"Andreas Jaeger","email":"aj@suse.com","date":"2019-08-08 07:03:44.000000000","tz":120},"committer":{"name":"Andreas Jaeger","email":"aj@suse.com","date":"2019-08-08 07:03:44.000000000","tz":120},"subject":"Bump ansible-lint to version 4","message":"Bump ansible-lint to version 4\n\nUpdate for newer ansible support.\n\nChange-Id: I982f20385889bb1d9721b18cca179e49e39b903a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/58fb09ff39c63f83970a80b26877a862ee0723e4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/58fb09ff39c63f83970a80b26877a862ee0723e4"}]},"branch":"refs/heads/master"},"4805bceecf7a99cc2ff967781ad47a628ae0b0b6":{"kind":"REWORK","_number":2,"created":"2019-08-08 07:10:00.000000000","uploader":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"ref":"refs/changes/55/675255/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/55/675255/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/55/675255/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/55/675255/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/55/675255/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/55/675255/2"}}},"commit":{"parents":[{"commit":"4972c5fc8f9209140a5ef16b581deaa44b426af2","subject":"Merge \"projects: add vexxhost/ansible-role-openmanage\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/4972c5fc8f9209140a5ef16b581deaa44b426af2"}]}],"author":{"name":"Andreas Jaeger","email":"aj@suse.com","date":"2019-08-08 07:03:44.000000000","tz":120},"committer":{"name":"Andreas Jaeger","email":"aj@suse.com","date":"2019-08-08 07:09:20.000000000","tz":120},"subject":"Bump ansible-lint to version 4","message":"Bump ansible-lint to version 4\n\nUpdate for newer ansible support.\nDisable test 204, we do not want to enforce a max line limit.\n\nChange-Id: I982f20385889bb1d9721b18cca179e49e39b903a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/4805bceecf7a99cc2ff967781ad47a628ae0b0b6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/4805bceecf7a99cc2ff967781ad47a628ae0b0b6"}]},"branch":"refs/heads/master"},"b4067e4aac6acc5162c6ebb22048ea2b3725fb37":{"kind":"REWORK","_number":3,"created":"2019-08-08 07:43:56.000000000","uploader":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"ref":"refs/changes/55/675255/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/55/675255/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/55/675255/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/55/675255/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/55/675255/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/55/675255/3"}}},"commit":{"parents":[{"commit":"4972c5fc8f9209140a5ef16b581deaa44b426af2","subject":"Merge \"projects: add vexxhost/ansible-role-openmanage\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/4972c5fc8f9209140a5ef16b581deaa44b426af2"}]}],"author":{"name":"Andreas Jaeger","email":"aj@suse.com","date":"2019-08-08 07:03:44.000000000","tz":120},"committer":{"name":"Andreas Jaeger","email":"aj@suse.com","date":"2019-08-08 07:38:07.000000000","tz":120},"subject":"Bump ansible-lint to version 4","message":"Bump ansible-lint to version 4\n\nUpdate for newer ansible support.\n\nDisable test 204, we do not want to enforce a max line limit.\nDisable test 306 for pipefail.\nCopying over the narrative from opendev/system-config for these.\n\nFix some missing spaces around {{ and }}.\n\nDepends-On: https://review.opendev.org/667698\nDepends-On: https://review.opendev.org/675260\nChange-Id: I982f20385889bb1d9721b18cca179e49e39b903a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/b4067e4aac6acc5162c6ebb22048ea2b3725fb37"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/b4067e4aac6acc5162c6ebb22048ea2b3725fb37"}]},"branch":"refs/heads/master"},"3229c234442dccc777b50340b4e36d756022c950":{"kind":"REWORK","_number":4,"created":"2019-08-08 08:05:26.000000000","uploader":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"ref":"refs/changes/55/675255/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/55/675255/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/55/675255/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/55/675255/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/55/675255/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/55/675255/4"}}},"commit":{"parents":[{"commit":"4972c5fc8f9209140a5ef16b581deaa44b426af2","subject":"Merge \"projects: add vexxhost/ansible-role-openmanage\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/4972c5fc8f9209140a5ef16b581deaa44b426af2"}]}],"author":{"name":"Andreas Jaeger","email":"aj@suse.com","date":"2019-08-08 07:03:44.000000000","tz":120},"committer":{"name":"Andreas Jaeger","email":"aj@suse.com","date":"2019-08-08 08:04:34.000000000","tz":120},"subject":"Bump ansible-lint to version 4","message":"Bump ansible-lint to version 4\n\nUpdate for newer ansible support.\n\nDisable test 204, we do not want to enforce a max line limit.\nDisable test 306 for pipefail.\nCopying over the narrative from opendev/system-config for these.\n\nDisable 701 - we do not use galaxy so far, so no need to warn about\nmissing galaxy_info in roles/submit-subunit-jobs/meta/main.yaml.\n\nFix some missing spaces around {{ and }}.\n\nDepends-On: https://review.opendev.org/667698\nDepends-On: https://review.opendev.org/675260\nChange-Id: I982f20385889bb1d9721b18cca179e49e39b903a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/3229c234442dccc777b50340b4e36d756022c950"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/3229c234442dccc777b50340b4e36d756022c950"}]},"branch":"refs/heads/master"},"62a1fa98ae6b40299e72c6ed730e8fe55a67886d":{"kind":"REWORK","_number":5,"created":"2019-08-08 08:06:32.000000000","uploader":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"ref":"refs/changes/55/675255/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/55/675255/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/55/675255/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/55/675255/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/55/675255/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/55/675255/5"}}},"commit":{"parents":[{"commit":"4972c5fc8f9209140a5ef16b581deaa44b426af2","subject":"Merge \"projects: add vexxhost/ansible-role-openmanage\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/4972c5fc8f9209140a5ef16b581deaa44b426af2"}]}],"author":{"name":"Andreas Jaeger","email":"aj@suse.com","date":"2019-08-08 07:03:44.000000000","tz":120},"committer":{"name":"Andreas Jaeger","email":"aj@suse.com","date":"2019-08-08 08:06:24.000000000","tz":120},"subject":"Bump ansible-lint to version 4","message":"Bump ansible-lint to version 4\n\nUpdate for newer ansible support.\n\nDisable test 204, we do not want to enforce a max line limit.\nDisable test 306 for pipefail.\nCopying over the narrative from opendev/system-config for these.\n\nDisable 701 - we do not use galaxy so far, so no need to warn about\nmissing galaxy_info in roles/submit-subunit-jobs/meta/main.yaml.\n\nFix some missing spaces around {{ and }}.\n\nDepends-On: https://review.opendev.org/667698\nDepends-On: https://review.opendev.org/675260\nChange-Id: I982f20385889bb1d9721b18cca179e49e39b903a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/62a1fa98ae6b40299e72c6ed730e8fe55a67886d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/62a1fa98ae6b40299e72c6ed730e8fe55a67886d"}]},"branch":"refs/heads/master"},"6571b7671bf78b4f46ffdec89d1a5be1ee8644e2":{"kind":"REWORK","_number":6,"created":"2019-08-09 04:20:50.000000000","uploader":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"ref":"refs/changes/55/675255/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/55/675255/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/55/675255/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/55/675255/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/55/675255/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/55/675255/6"}}},"commit":{"parents":[{"commit":"4972c5fc8f9209140a5ef16b581deaa44b426af2","subject":"Merge \"projects: add vexxhost/ansible-role-openmanage\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/4972c5fc8f9209140a5ef16b581deaa44b426af2"}]}],"author":{"name":"Andreas Jaeger","email":"aj@suse.com","date":"2019-08-08 07:03:44.000000000","tz":120},"committer":{"name":"Andreas Jaeger","email":"aj@suse.com","date":"2019-08-09 04:20:13.000000000","tz":120},"subject":"Bump ansible-lint to version 4","message":"Bump ansible-lint to version 4\n\nUpdate for newer ansible support.\n\nDisable test 204, we do not want to enforce a max line limit.\nDisable test 306 for pipefail.\nCopying over the narrative from opendev/system-config for these.\n\nDisable 701 - we do not use galaxy so far, so no need to warn about\nmissing galaxy_info in roles/submit-subunit-jobs/meta/main.yaml.\n\nDisable 403, we do install latest pip in some cases.\n\nFix some missing spaces around {{ and }}.\n\nDepends-On: https://review.opendev.org/667698\nDepends-On: https://review.opendev.org/675260\nChange-Id: I982f20385889bb1d9721b18cca179e49e39b903a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/6571b7671bf78b4f46ffdec89d1a5be1ee8644e2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/6571b7671bf78b4f46ffdec89d1a5be1ee8644e2"}]},"branch":"refs/heads/master"},"d1d284175df0ace5ee3e5be56684f861364231e1":{"kind":"REWORK","_number":7,"created":"2019-08-09 04:21:54.000000000","uploader":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"ref":"refs/changes/55/675255/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/55/675255/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/55/675255/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/55/675255/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/55/675255/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/55/675255/7"}}},"commit":{"parents":[{"commit":"4972c5fc8f9209140a5ef16b581deaa44b426af2","subject":"Merge \"projects: add vexxhost/ansible-role-openmanage\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/4972c5fc8f9209140a5ef16b581deaa44b426af2"}]}],"author":{"name":"Andreas Jaeger","email":"aj@suse.com","date":"2019-08-08 07:03:44.000000000","tz":120},"committer":{"name":"Andreas Jaeger","email":"aj@suse.com","date":"2019-08-09 04:21:45.000000000","tz":120},"subject":"Bump ansible-lint to version 4","message":"Bump ansible-lint to version 4\n\nUpdate for newer ansible support.\n\nDisable test 204, we do not want to enforce a max line limit.\nDisable test 306 for pipefail.\nCopying over the narrative from opendev/system-config for these.\n\nDisable 701 - we do not use galaxy so far, so no need to warn about\nmissing galaxy_info in roles/submit-subunit-jobs/meta/main.yaml.\n\nDisable 403, we do install latest pip in some cases.\n\nFix some missing spaces around {{ and }}.\n\nDepends-On: https://review.opendev.org/667698\nDepends-On: https://review.opendev.org/675260\nChange-Id: I982f20385889bb1d9721b18cca179e49e39b903a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/d1d284175df0ace5ee3e5be56684f861364231e1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/d1d284175df0ace5ee3e5be56684f861364231e1"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
