)]}'
{"id":"openstack%2Fproject-config~675641","triplet_id":"openstack%2Fproject-config~master~Idacbb57867ea37ce40963f47e6f0c25ba9e033ef","project":"openstack/project-config","branch":"master","topic":"swift-logs","hashtags":[],"change_id":"Idacbb57867ea37ce40963f47e6f0c25ba9e033ef","subject":"Report the build page url in Zuul","status":"MERGED","created":"2019-08-09 15:20:18.000000000","updated":"2019-08-12 15:34:22.000000000","submitted":"2019-08-12 15:34:22.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"675641-1565624062785-ff059232","meta_rev_id":"1e28259260b244835123698c9099b2a7892a7edd","_number":675641,"virtual_id_number":675641,"owner":{"_account_id":1,"name":"James E. Blair","email":"jim@acmegating.com","username":"corvus"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":1,"name":"James E. Blair","email":"jim@acmegating.com","username":"corvus"},{"value":0,"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"value":0,"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":2,"date":"2019-08-12 15:34:22.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"all":[{"value":0,"_account_id":1,"name":"James E. Blair","email":"jim@acmegating.com","username":"corvus"},{"value":2,"date":"2019-08-09 15:21:23.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"value":2,"date":"2019-08-09 18:07:29.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":1,"name":"James E. Blair","email":"jim@acmegating.com","username":"corvus"},"all":[{"value":1,"date":"2019-08-12 15:20:54.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":1,"name":"James E. Blair","email":"jim@acmegating.com","username":"corvus"},{"value":0,"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"value":0,"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1,"name":"James E. Blair","email":"jim@acmegating.com","username":"corvus"},{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-08-09 15:21:23.000000000","updated_by":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"reviewer":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"state":"REVIEWER"},{"updated":"2019-08-09 18:07:29.000000000","updated_by":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"reviewer":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"state":"REVIEWER"},{"updated":"2019-08-12 15:34:22.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"d86a8f0970e32ce8ad86222d9930539982707610","author":{"_account_id":1,"name":"James E. Blair","email":"jim@acmegating.com","username":"corvus"},"date":"2019-08-09 15:20:18.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"765c9d7cf2040eeda5be534ffcf258bac3af8542","author":{"_account_id":1,"name":"James E. Blair","email":"jim@acmegating.com","username":"corvus"},"date":"2019-08-09 15:20:46.000000000","message":"Patch Set 1: Workflow-1\n\nThis should wait until Monday.","accounts_in_message":[],"_revision_number":1},{"id":"bd5d8962a6eaa6a16a5d67d0e68691ed386a9931","author":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"date":"2019-08-09 15:21:23.000000000","message":"Patch Set 1: Code-Review+2\n\nLGTM for when we merge this monday.","accounts_in_message":[],"_revision_number":1},{"id":"e737bfecfb493e980ea0da2996a2165dade1faa3","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-09 15:28:01.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-zuul-jobs-linters https://logs.opendev.org/41/675641/1/check/openstack-zuul-jobs-linters/c6d7153/ : SUCCESS in 6m 26s\n- project-config-gerrit https://logs.opendev.org/41/675641/1/check/project-config-gerrit/6cffe54/ : SUCCESS in 4m 22s","accounts_in_message":[],"_revision_number":1},{"id":"c974b025de4976d39a8d1759b3698e8daa61f4dd","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2019-08-09 18:07:29.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"eac045ea23d13500439fc1afec482ee5479bb8a8","author":{"_account_id":1,"name":"James E. Blair","email":"jim@acmegating.com","username":"corvus"},"date":"2019-08-12 15:20:54.000000000","message":"Patch Set 1: Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"d2a2ad704da2431972c1b6c3a9c99b07e2621b13","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-12 15:21:07.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"7bae8d934eaedec0feb2dc6a5185c118c080372f","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-12 15:34:22.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-zuul-jobs-linters https://logs.opendev.org/41/675641/1/gate/openstack-zuul-jobs-linters/71efd33/ : SUCCESS in 7m 03s\n- project-config-gerrit https://logs.opendev.org/41/675641/1/gate/project-config-gerrit/3eb8290/ : SUCCESS in 4m 06s","accounts_in_message":[],"_revision_number":1},{"id":"21f28a2a071c0a53343359da879e168bc1ff40a0","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-12 15:34:22.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"44e5a21f4435ada5a6da4b763808a3c39dd8936e","revisions":{"44e5a21f4435ada5a6da4b763808a3c39dd8936e":{"kind":"REWORK","_number":1,"created":"2019-08-09 15:20:18.000000000","uploader":{"_account_id":1,"name":"James E. Blair","email":"jim@acmegating.com","username":"corvus"},"ref":"refs/changes/41/675641/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/41/675641/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/41/675641/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/41/675641/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/41/675641/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/41/675641/1"}}},"commit":{"parents":[{"commit":"0f977df14d0514d84ef1c2b00ce975b68521e77a","subject":"starlingx: add zuul-jobs repo","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/0f977df14d0514d84ef1c2b00ce975b68521e77a"}]}],"author":{"name":"James E. Blair","email":"jeblair@redhat.com","date":"2019-08-09 15:17:54.000000000","tz":-420},"committer":{"name":"James E. Blair","email":"jeblair@redhat.com","date":"2019-08-09 15:17:54.000000000","tz":-420},"subject":"Report the build page url in Zuul","message":"Report the build page url in Zuul\n\nAt the conclusion of jobs, rather than reporting direct links to\nthe log archive locations to Gerrit, report a link to the build\npage for each build.  This will allow us to reduce the reliance\non OSLA and our log server and eventually move to storing logs\nin swift.\n\nChange-Id: Idacbb57867ea37ce40963f47e6f0c25ba9e033ef\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/44e5a21f4435ada5a6da4b763808a3c39dd8936e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/44e5a21f4435ada5a6da4b763808a3c39dd8936e"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
