)]}'
{"id":"openstack%2Fproject-config~676422","triplet_id":"openstack%2Fproject-config~master~I4f92ad7e43e3e5f6a745583b0d36fa684aec2b42","project":"openstack/project-config","branch":"master","hashtags":[],"change_id":"I4f92ad7e43e3e5f6a745583b0d36fa684aec2b42","subject":"projects: add openstack/ansible-role-uwsgi to zuul","status":"MERGED","created":"2019-08-14 14:25:15.000000000","updated":"2019-08-15 15:14:32.000000000","submitted":"2019-08-15 15:14:32.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"676422-1565882072017-e4a4d76b","meta_rev_id":"7901e933a9040aba00046d5115a1b4667bec4df0","_number":676422,"virtual_id_number":676422,"owner":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":2,"date":"2019-08-15 15:14:31.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"all":[{"value":2,"date":"2019-08-14 17:37:25.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2019-08-15 15:03:09.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"all":[{"value":0,"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2019-08-15 15:03:09.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-08-14 14:28:44.000000000","updated_by":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"reviewer":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"state":"REVIEWER"},{"updated":"2019-08-15 15:03:09.000000000","updated_by":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"reviewer":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"state":"REVIEWER"},{"updated":"2019-08-15 15:14:31.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"6289bc9322aab1ac6606fe0122591d7c8f1e1a9f","author":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"date":"2019-08-14 14:25:15.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"f0006620b30e3aab73038a35316d0d8f297e7f68","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2019-08-14 14:28:44.000000000","message":"Patch Set 1: Code-Review+2\n\nSorry for missing that in review...","accounts_in_message":[],"_revision_number":1},{"id":"67a33b3ad6088e673f5a1e8df4caf6d26a31144e","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-14 14:35:25.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/e291f68ea57043589f8023b4e6b39299 : SUCCESS in 7m 29s\n- project-config-gerrit https://zuul.opendev.org/t/openstack/build/5fc831494c5d4effa2e4c9acd0f03c14 : FAILURE in 4m 59s","accounts_in_message":[],"_revision_number":1},{"id":"4512e5d91bf8b3cd7b7301fb4c4c0503892b00ad","author":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"date":"2019-08-14 17:37:25.000000000","message":"Patch Set 2: Patch Set 1 was rebased","accounts_in_message":[],"_revision_number":2},{"id":"2dfb502df12b803fd693fe9fef9b063de223517c","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-14 17:48:22.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/280e53b7fa5e4b60b29b3b69c25869e2 : SUCCESS in 6m 24s\n- project-config-gerrit https://zuul.opendev.org/t/openstack/build/f184e5f0ce304ca598737828223afdab : SUCCESS in 4m 19s","accounts_in_message":[],"_revision_number":2},{"id":"e6fb75f596d371e1bcb46a4512e5b795decd9730","author":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"date":"2019-08-15 15:03:09.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"ee9a352ce437760fac5eb53e893f3e5e854c199e","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-15 15:03:21.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"dbe4bf14546511cf9b5ec145fc37a1662b99bd91","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-15 15:14:31.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/0f2c9caf4ba14bef9a466793821f20b1 : SUCCESS in 7m 35s\n- project-config-gerrit https://zuul.opendev.org/t/openstack/build/99983df27702454f868c69e97600ed62 : SUCCESS in 4m 35s","accounts_in_message":[],"_revision_number":2},{"id":"7901e933a9040aba00046d5115a1b4667bec4df0","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-15 15:14:32.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"0caf81c57ae9119ddcb702d250a1dd4823623fa0","revisions":{"43ab6e03759433a05f11cc5f0d0e3320a4c0aec6":{"kind":"REWORK","_number":1,"created":"2019-08-14 14:25:15.000000000","uploader":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"ref":"refs/changes/22/676422/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/22/676422/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/22/676422/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/22/676422/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/22/676422/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/22/676422/1"}}},"commit":{"parents":[{"commit":"7d6e534e438d9bde3da0e5c729b38924f390c033","subject":"projects: add openstack/ansible-plugin-container-connection","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/7d6e534e438d9bde3da0e5c729b38924f390c033"}]}],"author":{"name":"Mohammed Naser","email":"mnaser@vexxhost.com","date":"2019-08-14 14:18:53.000000000","tz":-240},"committer":{"name":"Mohammed Naser","email":"mnaser@vexxhost.com","date":"2019-08-14 14:25:08.000000000","tz":-240},"subject":"projects: add openstack/ansible-role-uwsgi to zuul","message":"projects: add openstack/ansible-role-uwsgi to zuul\n\nIt seems like we added it to Gerrit but forgot to add it to Zuul.\n\nChange-Id: I4f92ad7e43e3e5f6a745583b0d36fa684aec2b42\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/43ab6e03759433a05f11cc5f0d0e3320a4c0aec6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/43ab6e03759433a05f11cc5f0d0e3320a4c0aec6"}]},"branch":"refs/heads/master"},"0caf81c57ae9119ddcb702d250a1dd4823623fa0":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2019-08-14 17:37:25.000000000","uploader":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"ref":"refs/changes/22/676422/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/22/676422/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/22/676422/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/22/676422/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/22/676422/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/22/676422/2"}}},"commit":{"parents":[{"commit":"8beaa9ea225e5654ed4074ae71466a71f9b00eab","subject":"Enable Review Priority flag for Monasca","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/8beaa9ea225e5654ed4074ae71466a71f9b00eab"}]}],"author":{"name":"Mohammed Naser","email":"mnaser@vexxhost.com","date":"2019-08-14 14:18:53.000000000","tz":-240},"committer":{"name":"Mohammed Naser","email":"mnaser@vexxhost.com","date":"2019-08-14 17:37:25.000000000","tz":0},"subject":"projects: add openstack/ansible-role-uwsgi to zuul","message":"projects: add openstack/ansible-role-uwsgi to zuul\n\nIt seems like we added it to Gerrit but forgot to add it to Zuul.\n\nChange-Id: I4f92ad7e43e3e5f6a745583b0d36fa684aec2b42\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/0caf81c57ae9119ddcb702d250a1dd4823623fa0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/0caf81c57ae9119ddcb702d250a1dd4823623fa0"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
