)]}'
{"id":"openstack%2Fproject-config~676630","triplet_id":"openstack%2Fproject-config~master~I6ca0b538b68a1dacd1144398975608f01e18d7de","project":"openstack/project-config","branch":"master","topic":"promote-docs","hashtags":[],"change_id":"I6ca0b538b68a1dacd1144398975608f01e18d7de","subject":"Add promote api-ref/guide jobs","status":"MERGED","created":"2019-08-15 08:16:31.000000000","updated":"2019-08-19 06:23:41.000000000","submitted":"2019-08-19 06:23:41.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":4,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"676630-1566195821426-5b3f03e3","meta_rev_id":"e07446b04fc00fa4e448f8ff97ddfb0ae920dcc4","_number":676630,"virtual_id_number":676630,"owner":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":1,"name":"James E. Blair","email":"jim@acmegating.com","username":"corvus"},{"value":0,"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"value":0,"date":"2019-08-19 06:14:22.000000000","_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":2,"date":"2019-08-19 06:23:41.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},{"value":0,"_account_id":10607,"name":"Alexandra Settle","email":"a.settle@outlook.com","username":"asettle"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"all":[{"value":0,"_account_id":1,"name":"James E. Blair","email":"jim@acmegating.com","username":"corvus"},{"value":2,"date":"2019-08-16 16:49:27.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"value":0,"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2019-08-19 06:09:43.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},{"value":0,"_account_id":10607,"name":"Alexandra Settle","email":"a.settle@outlook.com","username":"asettle"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"all":[{"value":0,"_account_id":1,"name":"James E. Blair","email":"jim@acmegating.com","username":"corvus"},{"value":0,"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"value":0,"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2019-08-19 06:13:20.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},{"value":0,"_account_id":10607,"name":"Alexandra Settle","email":"a.settle@outlook.com","username":"asettle"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1,"name":"James E. Blair","email":"jim@acmegating.com","username":"corvus"},{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},{"_account_id":10607,"name":"Alexandra Settle","email":"a.settle@outlook.com","username":"asettle"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-08-15 10:29:59.000000000","updated_by":{"_account_id":10607,"name":"Alexandra Settle","email":"a.settle@outlook.com","username":"asettle"},"reviewer":{"_account_id":10607,"name":"Alexandra Settle","email":"a.settle@outlook.com","username":"asettle"},"state":"REVIEWER"},{"updated":"2019-08-16 16:07:53.000000000","updated_by":{"_account_id":1,"name":"James E. Blair","email":"jim@acmegating.com","username":"corvus"},"reviewer":{"_account_id":1,"name":"James E. Blair","email":"jim@acmegating.com","username":"corvus"},"state":"REVIEWER"},{"updated":"2019-08-16 16:49:27.000000000","updated_by":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"reviewer":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"state":"REVIEWER"},{"updated":"2019-08-19 06:13:20.000000000","updated_by":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"reviewer":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"state":"REVIEWER"},{"updated":"2019-08-19 06:23:41.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"e6f4908e547cf0e0944dbe39649c91dbe3c66512","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2019-08-15 08:16:31.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"0821ee4873d4d0b9b38313c4e998ee95183a4b79","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2019-08-15 08:21:57.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"4c853d0507dba72d44153c06bb3082ac0e8c04c4","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2019-08-15 08:34:10.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"173511a5b214ce873f209878f9734e8d9ca377b2","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-15 08:53:33.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/5283e030dbae4576ae57228060713186 : SUCCESS in 6m 53s","accounts_in_message":[],"_revision_number":3},{"id":"86e9c0367518a8afe3038f0b8c1ccc9d1cd8bd72","author":{"_account_id":10607,"name":"Alexandra Settle","email":"a.settle@outlook.com","username":"asettle"},"date":"2019-08-15 10:29:59.000000000","message":"Patch Set 3: Code-Review+1","accounts_in_message":[],"_revision_number":3},{"id":"9bd0c717b3b5da432e3ad0c779aca74c6bb2a778","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2019-08-15 17:11:02.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"3ce14431ad0ee71562a08ab7aa25ed3d38a394d4","author":{"_account_id":1,"name":"James E. Blair","email":"jim@acmegating.com","username":"corvus"},"date":"2019-08-15 17:12:58.000000000","message":"Patch Set 4: Code-Review+2","accounts_in_message":[],"_revision_number":4},{"id":"49095be835809ecf6b55dc3273e1664a28c35d04","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-15 17:22:15.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/4dffb6a358b04693ade27b0f02118ab1 : FAILURE in 8m 05s","accounts_in_message":[],"_revision_number":4},{"id":"76de77b5937b1f69519001e51efb1cc438b1cf6f","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2019-08-15 17:23:38.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"f91608159f52a50a6f8144ad91a978f963702920","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-15 17:37:27.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/654b7832233045f08fbc2605593a4caf : FAILURE in 7m 02s","accounts_in_message":[],"_revision_number":5},{"id":"554946c5d5a2308f4f90a431717f29f25b738392","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2019-08-15 17:54:21.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"cdf22e47353de267e9407d380b7c4b82cce44496","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-15 18:08:46.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/f67fdd22234d4bc682082775228e4a91 : SUCCESS in 7m 12s","accounts_in_message":[],"_revision_number":6},{"id":"9b0f9c36a8a7ab7c194c7acbccc7fdb66727604d","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2019-08-16 15:54:04.000000000","message":"Uploaded patch set 7.","accounts_in_message":[],"_revision_number":7},{"id":"ae7501dac9d6e33464b44cafb3395bc7591721ae","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-16 16:01:11.000000000","message":"Patch Set 7: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/9ee33a2e9a524a6c8951da803ed14c56 : SUCCESS in 6m 03s","accounts_in_message":[],"_revision_number":7},{"id":"1238168e19c3f739a86bfec2cfc06913f0d40710","author":{"_account_id":1,"name":"James E. Blair","email":"jim@acmegating.com","username":"corvus"},"date":"2019-08-16 16:07:53.000000000","message":"Patch Set 7:\n\n(2 comments)\n\nQuestion inline.\n\nWe could do this, but we could instead drop final:true and make this a parent job that is inherited from in openstack-zuul-jobs, which will put it next to the job that builds the docs and make it easier to keep the file matchers in sync.\n\nWe just need to make sure that the playbook itself will only work on appropriate projects (ie, if it can\u0027t find a service entry for the current project, it should fail) and that it doesn\u0027t act on anything that someone could change by adding a pre-playbook or setting a variable.","accounts_in_message":[],"_revision_number":7},{"id":"f443b7d9b5e1efd25c02a219975ae0a12e71849a","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2019-08-16 16:36:22.000000000","message":"Patch Set 7:\n\n(2 comments)","accounts_in_message":[],"_revision_number":7},{"id":"4e89daa7ea75de8784489c46e88e84ec9a23b627","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2019-08-16 16:38:48.000000000","message":"Uploaded patch set 8.","accounts_in_message":[],"_revision_number":8},{"id":"b9155eb0f3d960818298baaf88c927c1a884cb91","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-16 16:46:23.000000000","message":"Patch Set 8: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/a0ff9834e0e142c98c14aa550b395901 : SUCCESS in 7m 00s","accounts_in_message":[],"_revision_number":8},{"id":"022d7a299d1e0830535ead4c41a38e02aff892e8","author":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"date":"2019-08-16 16:49:27.000000000","message":"Patch Set 8: Code-Review+2","accounts_in_message":[],"_revision_number":8},{"id":"226848a9ce3a10f0b75192822ae662ac09a1e52b","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2019-08-16 17:58:36.000000000","message":"Topic changed from promote-openstack-manuals to promote-docs","accounts_in_message":[],"_revision_number":8},{"id":"762c6f6709410578578f311054a0c96d851169d6","author":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"date":"2019-08-19 06:09:43.000000000","message":"Patch Set 8: Code-Review+2","accounts_in_message":[],"_revision_number":8},{"id":"70ba72230fdecdc9502370f500e57295f31435ff","author":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"date":"2019-08-19 06:13:20.000000000","message":"Patch Set 8: Workflow+1\n\ni think jeblairs comments could be implemented by moving things in a follow-up, if required","accounts_in_message":[],"_revision_number":8},{"id":"a3706ce2a97fe7a0e683b58807dc178494fcf295","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-19 06:13:39.000000000","message":"Patch Set 8: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":8},{"id":"e314dd83ee4ad3b4fac66106afa93d847ca91406","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2019-08-19 06:14:22.000000000","message":"Patch Set 8:\n\nI discussed with Jim on IRC and updated the files list instead in openstack-zuul-jobs, see needed-by.","accounts_in_message":[],"_revision_number":8},{"id":"03b90c97d8833bf8235743a541c63f01a58b77a6","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-19 06:23:41.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":8},{"id":"e07446b04fc00fa4e448f8ff97ddfb0ae920dcc4","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-19 06:23:41.000000000","message":"Patch Set 8: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/c45d76c8b8e846dda3e2ae0ae4f64835 : SUCCESS in 8m 02s","accounts_in_message":[],"_revision_number":8}],"current_revision_number":8,"current_revision":"501ce8027543cb68ef424998632287cc262d0cf6","revisions":{"f25bb7081128b7b2850cf24f45c577d2a5a7ec37":{"kind":"REWORK","_number":1,"created":"2019-08-15 08:16:31.000000000","uploader":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"ref":"refs/changes/30/676630/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/30/676630/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/30/676630/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/30/676630/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/30/676630/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/30/676630/1"}}},"commit":{"parents":[{"commit":"d0ab69070af40e6b597a260673e6b755c12f814a","subject":"Merge \"Enable Review Priority flag for Monasca\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/d0ab69070af40e6b597a260673e6b755c12f814a"}]}],"author":{"name":"Andreas Jaeger","email":"aj@suse.com","date":"2019-08-15 08:15:07.000000000","tz":120},"committer":{"name":"Andreas Jaeger","email":"aj@suse.com","date":"2019-08-15 08:15:07.000000000","tz":120},"subject":"Add promote api-ref/guide jobs","message":"Add promote api-ref/guide jobs\n\nAdd promote jobs for api-ref and api-guide jobs.\n\nRelated change: https://review.opendev.org/676624\n\nChange-Id: I6ca0b538b68a1dacd1144398975608f01e18d7de\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/f25bb7081128b7b2850cf24f45c577d2a5a7ec37"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/f25bb7081128b7b2850cf24f45c577d2a5a7ec37"}]},"branch":"refs/heads/master"},"be0cfc70dac1ca6c8e6231b5f80ebe6c8a8d882e":{"kind":"REWORK","_number":2,"created":"2019-08-15 08:21:57.000000000","uploader":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"ref":"refs/changes/30/676630/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/30/676630/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/30/676630/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/30/676630/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/30/676630/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/30/676630/2"}}},"commit":{"parents":[{"commit":"d0ab69070af40e6b597a260673e6b755c12f814a","subject":"Merge \"Enable Review Priority flag for Monasca\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/d0ab69070af40e6b597a260673e6b755c12f814a"}]}],"author":{"name":"Andreas Jaeger","email":"aj@suse.com","date":"2019-08-15 08:15:07.000000000","tz":120},"committer":{"name":"Andreas Jaeger","email":"aj@suse.com","date":"2019-08-15 08:21:46.000000000","tz":120},"subject":"Add promote api-ref/guide jobs","message":"Add promote api-ref/guide jobs\n\nAdd promote jobs for api-ref and api-guide jobs.\n\nRelated change: https://review.opendev.org/676624\n\nNeeded-By: https://review.opendev.org/676631\nChange-Id: I6ca0b538b68a1dacd1144398975608f01e18d7de\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/be0cfc70dac1ca6c8e6231b5f80ebe6c8a8d882e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/be0cfc70dac1ca6c8e6231b5f80ebe6c8a8d882e"}]},"branch":"refs/heads/master"},"f500a88a34652e8f8284c39886e36d1aaac87f09":{"kind":"REWORK","_number":3,"created":"2019-08-15 08:34:10.000000000","uploader":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"ref":"refs/changes/30/676630/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/30/676630/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/30/676630/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/30/676630/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/30/676630/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/30/676630/3"}}},"commit":{"parents":[{"commit":"15b93a404b1d8111e0e1dca3d4fe50a79f66e158","subject":"Add promote-openstack-manuals jobs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/15b93a404b1d8111e0e1dca3d4fe50a79f66e158"}]}],"author":{"name":"Andreas Jaeger","email":"aj@suse.com","date":"2019-08-15 08:15:07.000000000","tz":120},"committer":{"name":"Andreas Jaeger","email":"aj@suse.com","date":"2019-08-15 08:33:57.000000000","tz":120},"subject":"WIP: Add promote api-ref/guide jobs","message":"WIP: Add promote api-ref/guide jobs\n\nAdd promote jobs for api-ref and api-guide jobs.\n\nWIP: This needs further change for the publishing path since\nwe need to compute the path:\n    - role: set-service-type-data-fact\n    - role: upload-afs\n      afs_target: \"{{ afs.path }}/{{ api_path }}/{{ service_type_data.service_type }}\"\n\nRelated change: https://review.opendev.org/676624\n\nNeeded-By: https://review.opendev.org/676631\nChange-Id: I6ca0b538b68a1dacd1144398975608f01e18d7de\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/f500a88a34652e8f8284c39886e36d1aaac87f09"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/f500a88a34652e8f8284c39886e36d1aaac87f09"}]},"branch":"refs/heads/master"},"7d0015baaedfdc86dd896e1967b0f93c48919ee4":{"kind":"REWORK","_number":4,"created":"2019-08-15 17:11:02.000000000","uploader":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"ref":"refs/changes/30/676630/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/30/676630/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/30/676630/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/30/676630/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/30/676630/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/30/676630/4"}}},"commit":{"parents":[{"commit":"e3cd1b09d16fd260443812d2aad94b01cfaf7fba","subject":"Merge \"Add promote-openstack-manuals jobs\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/e3cd1b09d16fd260443812d2aad94b01cfaf7fba"}]}],"author":{"name":"Andreas Jaeger","email":"aj@suse.com","date":"2019-08-15 08:15:07.000000000","tz":120},"committer":{"name":"Andreas Jaeger","email":"aj@suse.com","date":"2019-08-15 17:10:44.000000000","tz":120},"subject":"Add promote api-ref/guide jobs","message":"Add promote api-ref/guide jobs\n\nAdd promote jobs for api-ref and api-guide jobs.\n\nThis adds a promote playbook since we need to calculate teh service\ntype, for example for nova we publish to compute.\n\nRelated change: https://review.opendev.org/676624\n\nNeeded-By: https://review.opendev.org/676631\nChange-Id: I6ca0b538b68a1dacd1144398975608f01e18d7de\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/7d0015baaedfdc86dd896e1967b0f93c48919ee4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/7d0015baaedfdc86dd896e1967b0f93c48919ee4"}]},"branch":"refs/heads/master"},"bbdd161bfcd94708bd44fbf40b9438c5ebc73b0f":{"kind":"REWORK","_number":5,"created":"2019-08-15 17:23:38.000000000","uploader":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"ref":"refs/changes/30/676630/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/30/676630/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/30/676630/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/30/676630/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/30/676630/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/30/676630/5"}}},"commit":{"parents":[{"commit":"e3cd1b09d16fd260443812d2aad94b01cfaf7fba","subject":"Merge \"Add promote-openstack-manuals jobs\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/e3cd1b09d16fd260443812d2aad94b01cfaf7fba"}]}],"author":{"name":"Andreas Jaeger","email":"aj@suse.com","date":"2019-08-15 08:15:07.000000000","tz":120},"committer":{"name":"Andreas Jaeger","email":"aj@suse.com","date":"2019-08-15 17:23:29.000000000","tz":120},"subject":"Add promote api-ref/guide jobs","message":"Add promote api-ref/guide jobs\n\nAdd promote jobs for api-ref and api-guide jobs.\n\nThis adds a promote playbook since we need to calculate teh service\ntype, for example for nova we publish to compute.\n\nRelated change: https://review.opendev.org/676624\n\nNeeded-By: https://review.opendev.org/676631\nChange-Id: I6ca0b538b68a1dacd1144398975608f01e18d7de\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/bbdd161bfcd94708bd44fbf40b9438c5ebc73b0f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/bbdd161bfcd94708bd44fbf40b9438c5ebc73b0f"}]},"branch":"refs/heads/master"},"1debd185b7e906b5ec6ccdf3da04224620bbd785":{"kind":"REWORK","_number":6,"created":"2019-08-15 17:54:21.000000000","uploader":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"ref":"refs/changes/30/676630/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/30/676630/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/30/676630/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/30/676630/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/30/676630/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/30/676630/6"}}},"commit":{"parents":[{"commit":"e3cd1b09d16fd260443812d2aad94b01cfaf7fba","subject":"Merge \"Add promote-openstack-manuals jobs\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/e3cd1b09d16fd260443812d2aad94b01cfaf7fba"}]}],"author":{"name":"Andreas Jaeger","email":"aj@suse.com","date":"2019-08-15 08:15:07.000000000","tz":120},"committer":{"name":"Andreas Jaeger","email":"aj@suse.com","date":"2019-08-15 17:54:11.000000000","tz":120},"subject":"Add promote api-ref/guide jobs","message":"Add promote api-ref/guide jobs\n\nAdd promote jobs for api-ref and api-guide jobs.\n\nThis adds a promote playbook since we need to calculate teh service\ntype, for example for nova we publish to compute.\n\nRelated change: https://review.opendev.org/676624\n\nNeeded-By: https://review.opendev.org/676631\nChange-Id: I6ca0b538b68a1dacd1144398975608f01e18d7de\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/1debd185b7e906b5ec6ccdf3da04224620bbd785"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/1debd185b7e906b5ec6ccdf3da04224620bbd785"}]},"branch":"refs/heads/master"},"879e5536322dd31dc23e93241d5905045ff23864":{"kind":"REWORK","_number":7,"created":"2019-08-16 15:54:04.000000000","uploader":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"ref":"refs/changes/30/676630/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/30/676630/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/30/676630/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/30/676630/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/30/676630/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/30/676630/7"}}},"commit":{"parents":[{"commit":"e3cd1b09d16fd260443812d2aad94b01cfaf7fba","subject":"Merge \"Add promote-openstack-manuals jobs\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/e3cd1b09d16fd260443812d2aad94b01cfaf7fba"}]}],"author":{"name":"Andreas Jaeger","email":"aj@suse.com","date":"2019-08-15 08:15:07.000000000","tz":120},"committer":{"name":"Andreas Jaeger","email":"aj@suse.com","date":"2019-08-16 15:53:39.000000000","tz":120},"subject":"Add promote api-ref/guide jobs","message":"Add promote api-ref/guide jobs\n\nAdd promote jobs for api-ref and api-guide jobs.\n\nThis adds a promote playbook since we need to calculate the service\ntype, for example for nova we publish to compute.\n\nRelated change: https://review.opendev.org/676624\n\nNeeded-By: https://review.opendev.org/676631\nChange-Id: I6ca0b538b68a1dacd1144398975608f01e18d7de\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/879e5536322dd31dc23e93241d5905045ff23864"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/879e5536322dd31dc23e93241d5905045ff23864"}]},"branch":"refs/heads/master"},"501ce8027543cb68ef424998632287cc262d0cf6":{"kind":"REWORK","_number":8,"created":"2019-08-16 16:38:48.000000000","uploader":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"ref":"refs/changes/30/676630/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/30/676630/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/30/676630/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/30/676630/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/30/676630/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/30/676630/8"}}},"commit":{"parents":[{"commit":"e3cd1b09d16fd260443812d2aad94b01cfaf7fba","subject":"Merge \"Add promote-openstack-manuals jobs\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/e3cd1b09d16fd260443812d2aad94b01cfaf7fba"}]}],"author":{"name":"Andreas Jaeger","email":"aj@suse.com","date":"2019-08-15 08:15:07.000000000","tz":120},"committer":{"name":"Andreas Jaeger","email":"aj@suse.com","date":"2019-08-16 16:38:38.000000000","tz":120},"subject":"Add promote api-ref/guide jobs","message":"Add promote api-ref/guide jobs\n\nAdd promote jobs for api-ref and api-guide jobs.\n\nThis adds a promote playbook since we need to calculate the service\ntype, for example for nova we publish to compute.\n\nRelated change: https://review.opendev.org/676624\n\nNeeded-By: https://review.opendev.org/676631\nChange-Id: I6ca0b538b68a1dacd1144398975608f01e18d7de\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/501ce8027543cb68ef424998632287cc262d0cf6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/501ce8027543cb68ef424998632287cc262d0cf6"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
