)]}'
{"id":"openstack%2Fproject-config~676697","triplet_id":"openstack%2Fproject-config~master~I587a70f620bcfb1069508f69df07317d7194025c","project":"openstack/project-config","branch":"master","hashtags":[],"change_id":"I587a70f620bcfb1069508f69df07317d7194025c","subject":"Set default-ansible-version to 2.7 for all tenant (minus zuul)","status":"MERGED","created":"2019-08-15 13:45:40.000000000","updated":"2019-12-10 19:10:33.000000000","submitted":"2019-12-10 19:10:33.000000000","submitter":{"_account_id":1,"name":"James E. Blair","email":"jim@acmegating.com","username":"corvus"},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"676697-1576005033524-1e73c108","meta_rev_id":"fccb32019314b68edd0f2deac815709d98e7b1ac","_number":676697,"virtual_id_number":676697,"owner":{"_account_id":4162,"name":"Paul Belanger","email":"pabelanger@redhat.com","username":"pabelanger"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":1,"name":"James E. Blair","email":"jim@acmegating.com","username":"corvus"},"all":[{"value":2,"date":"2019-12-10 19:10:00.000000000","_account_id":1,"name":"James E. Blair","email":"jim@acmegating.com","username":"corvus"},{"value":0,"date":"2019-08-29 20:52:17.000000000","_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"value":0,"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":0,"date":"2019-12-10 19:10:18.000000000","permitted_voting_range":{"min":0,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":1,"name":"James E. Blair","email":"jim@acmegating.com","username":"corvus"},"all":[{"value":2,"date":"2019-12-10 19:10:00.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":1,"name":"James E. Blair","email":"jim@acmegating.com","username":"corvus"},{"value":0,"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"value":0,"date":"2019-09-20 17:51:04.000000000","permitted_voting_range":{"min":0,"max":2},"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2019-12-10 19:09:55.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":1,"name":"James E. Blair","email":"jim@acmegating.com","username":"corvus"},"all":[{"value":1,"date":"2019-12-10 19:10:00.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":1,"name":"James E. Blair","email":"jim@acmegating.com","username":"corvus"},{"value":0,"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"value":0,"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1,"name":"James E. Blair","email":"jim@acmegating.com","username":"corvus"},{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-08-29 20:52:17.000000000","updated_by":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"reviewer":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"state":"REVIEWER"},{"updated":"2019-09-20 17:51:04.000000000","updated_by":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"reviewer":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"state":"REVIEWER"},{"updated":"2019-12-10 19:09:55.000000000","updated_by":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"reviewer":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"state":"REVIEWER"},{"updated":"2019-12-10 19:10:00.000000000","updated_by":{"_account_id":1,"name":"James E. Blair","email":"jim@acmegating.com","username":"corvus"},"reviewer":{"_account_id":1,"name":"James E. Blair","email":"jim@acmegating.com","username":"corvus"},"state":"REVIEWER"},{"updated":"2019-12-10 19:10:18.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"f3f6193b695b3d905c42fdeef9c5d31ba44ad4d0","author":{"_account_id":4162,"name":"Paul Belanger","email":"pabelanger@redhat.com","username":"pabelanger"},"date":"2019-08-15 13:45:40.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"cc7a6795995a06a3621d6b2be8f23658190115b2","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-15 13:57:29.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/bea15820b4db44c7af903ad9747de985 : SUCCESS in 6m 03s\n- project-config-gerrit https://zuul.opendev.org/t/openstack/build/47ed203094064e83a879b1a1bef9817f : SUCCESS in 4m 34s","accounts_in_message":[],"_revision_number":1},{"id":"136c9276631d32c95ed86438218a46ec4cff45cc","author":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"date":"2019-08-29 20:52:17.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"1e2e401ae473a4ccb232463abfca02a02d3ecb10","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2019-09-20 17:42:21.000000000","message":"Patch Set 1: Code-Review+2\n\nAFAIK we use 2.7 already, so this is should be safe.","accounts_in_message":[],"_revision_number":1},{"id":"7452cdc1e6617c3fc617847c2577b156d447ca29","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2019-09-20 17:51:04.000000000","message":"Patch Set 1: -Code-Review\n\nFrom IRC:\n\n19:45 \u003cpabelanger\u003e if opendev base job is 2.8, and openstack parent to it, doesn\u0027t those jobs also move to 2.8 too?\n19:45 \u003cpabelanger\u003e or do we do split ansible-playbook pre run phase\n19:46 \u003cpabelanger\u003e doesn\u0027t look like it\n19:47 \u003cpabelanger\u003e is, in case of opendev, we might need to lock step min version of ansible across all tenants\n19:49 \u003cpabelanger\u003e Hmm, yah, I think we need to test that. Not sure what will happen if we bump opendev tenant, and say openstack stays on 2.7\n\nLet\u0027s test, please...","accounts_in_message":[],"_revision_number":1},{"id":"5065771fb31ffc12d2363af10acf6e080bf811d7","author":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"date":"2019-12-10 19:09:55.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"e7456b45dcf822073e865b4c12a35105cbe296b5","author":{"_account_id":1,"name":"James E. Blair","email":"jim@acmegating.com","username":"corvus"},"date":"2019-12-10 19:10:00.000000000","message":"Patch Set 1: Verified+2 Code-Review+2 Workflow+1\n\nForce-merging since all jobs are currently breaking.","accounts_in_message":[],"_revision_number":1},{"id":"2c80a1ef325e2aa5ddfec83e8b3bb50c2fdb20e6","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-10 19:10:18.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"fccb32019314b68edd0f2deac815709d98e7b1ac","author":{"_account_id":1,"name":"James E. Blair","email":"jim@acmegating.com","username":"corvus"},"date":"2019-12-10 19:10:33.000000000","message":"Change has been successfully merged by James E. Blair","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"dff7fccc130a7c64d422b402d85d1f8669640a39","revisions":{"dff7fccc130a7c64d422b402d85d1f8669640a39":{"kind":"REWORK","_number":1,"created":"2019-08-15 13:45:40.000000000","uploader":{"_account_id":4162,"name":"Paul Belanger","email":"pabelanger@redhat.com","username":"pabelanger"},"ref":"refs/changes/97/676697/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/97/676697/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/97/676697/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/97/676697/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/97/676697/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/97/676697/1"}}},"commit":{"parents":[{"commit":"d0ab69070af40e6b597a260673e6b755c12f814a","subject":"Merge \"Enable Review Priority flag for Monasca\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/d0ab69070af40e6b597a260673e6b755c12f814a"}]}],"author":{"name":"Paul Belanger","email":"pabelanger@redhat.com","date":"2019-08-15 13:43:54.000000000","tz":-240},"committer":{"name":"Paul Belanger","email":"pabelanger@redhat.com","date":"2019-08-15 13:44:12.000000000","tz":-240},"subject":"Set default-ansible-version to 2.7 for all tenant (minus zuul)","message":"Set default-ansible-version to 2.7 for all tenant (minus zuul)\n\nThis is to prevent to upgrade of ansible version used by jobs, if / when\nzuul project switches from ansible 2.7 to 2.8 for defaults.\n\nChange-Id: I587a70f620bcfb1069508f69df07317d7194025c\nSigned-off-by: Paul Belanger \u003cpabelanger@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/dff7fccc130a7c64d422b402d85d1f8669640a39"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/dff7fccc130a7c64d422b402d85d1f8669640a39"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
