)]}'
{"id":"openstack%2Fproject-config~677537","triplet_id":"openstack%2Fproject-config~master~I0da6c940b59a66feaf87e353ba53d473a35122f4","project":"openstack/project-config","branch":"master","topic":"promote-docs","hashtags":[],"change_id":"I0da6c940b59a66feaf87e353ba53d473a35122f4","subject":"Add more promote docs jobs","status":"MERGED","created":"2019-08-20 18:33:38.000000000","updated":"2019-08-20 19:07:48.000000000","submitted":"2019-08-20 19:07:48.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"677537-1566328068666-1b2da68d","meta_rev_id":"de5c32c72a01954a8f4e4412ee2c429849e1107c","_number":677537,"virtual_id_number":677537,"owner":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"value":2,"date":"2019-08-20 19:07:48.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"all":[{"value":2,"date":"2019-08-20 18:48:44.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2019-08-20 18:51:21.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"all":[{"value":0,"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2019-08-20 18:51:21.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-08-20 18:48:44.000000000","updated_by":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"reviewer":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"state":"REVIEWER"},{"updated":"2019-08-20 18:51:21.000000000","updated_by":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"reviewer":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"state":"REVIEWER"},{"updated":"2019-08-20 19:07:48.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"4f881768f38d3e5ff6390e5aad5d69f5ddfc716f","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2019-08-20 18:33:38.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"08853bc7191fa313b07c432501d327957ac3e793","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-20 18:33:57.000000000","message":"Patch Set 1: Verified-1\n\n(1 comment)\n\nZuul encountered a syntax error while parsing its configuration in the\nrepo openstack/project-config on branch master.  The error was:\n\n  expected a dictionary for dictionary value @ data[\u0027promote\u0027]\n\nThe error appears in the following project stanza:\n\n  project:\n      name: osf/openstackid\n      check:\n        jobs:\n          - legacy-laravel-openstackid-unittests\n          - tox-docs:\n              nodeset: ubuntu-bionic\n      gate:\n        jobs:\n          - legacy-laravel-openstackid-unittests\n          - tox-docs:\n              nodeset: ubuntu-bionic\n      post:\n        jobs:\n          - legacy-laravel-openstackid-release-branch\n      promote:\n        jobs\n          - publish-tox-docs-infra-bionic\n      release:\n        jobs:\n          - legacy-laravel-openstackid-release-master\n\n  in \"openstack/project-config/zuul.d/projects.yaml@master\", line 5206, column 3","accounts_in_message":[],"_revision_number":1},{"id":"1e83ba227fbd403acfd7c9ac77f771e368bb43d4","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2019-08-20 18:37:39.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"027f6bc01c63f4f119b5b5f43767a6b1fff5c06f","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2019-08-20 18:40:53.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"922d88a602767c073071adee0668a966dd95822e","author":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"date":"2019-08-20 18:48:44.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"9286ae6c4df0e2a770642d839791c645fb84b59d","author":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"date":"2019-08-20 18:51:21.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"d6ba9e547375fc0e450a145cb9344ef1f24da3c2","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-20 18:53:18.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/b3e7a325663a4ac0be4851030b50b528 : SUCCESS in 6m 54s","accounts_in_message":[],"_revision_number":3},{"id":"a85d20e02900d45667368cd39b62b956465f631e","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-20 18:53:36.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"7b6b05d0681b3531e14a9037638e6ce24db86e58","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-20 19:07:48.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/de77dd37648f49f0ad7f81c89ac2a71a : SUCCESS in 7m 53s","accounts_in_message":[],"_revision_number":3},{"id":"1a489e7f07adedf7bf84fee1b74806293d7029f8","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-20 19:07:48.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"6b89ce404df7d4a484d5c2972105ac1bc025be4e","revisions":{"d397dfa7adede871f39cec4ce9da14f5e4dfd858":{"kind":"REWORK","_number":1,"created":"2019-08-20 18:33:38.000000000","uploader":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"ref":"refs/changes/37/677537/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/37/677537/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/37/677537/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/37/677537/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/37/677537/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/37/677537/1"}}},"commit":{"parents":[{"commit":"291489176cecf7e087a3964680db21b037cfc672","subject":"Fix publish path for api-jobs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/291489176cecf7e087a3964680db21b037cfc672"}]}],"author":{"name":"Andreas Jaeger","email":"aj@suse.com","date":"2019-08-20 18:30:32.000000000","tz":120},"committer":{"name":"Andreas Jaeger","email":"aj@suse.com","date":"2019-08-20 18:32:46.000000000","tz":120},"subject":"Add more promote docs jobs","message":"Add more promote docs jobs\n\nAdd promote-openstack-tox-docs-direct job for publishing jobs without\n/latest.\n\nAdd promote-openstack-tox-docs-infra job for publishing infra repos.\nConvert osf/openstackid to use it and remove release job, the infra job\ndoes not handle releases extra.\n\nChange-Id: I0da6c940b59a66feaf87e353ba53d473a35122f4\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/d397dfa7adede871f39cec4ce9da14f5e4dfd858"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/d397dfa7adede871f39cec4ce9da14f5e4dfd858"}]},"branch":"refs/heads/master"},"da91ffe986ed89b9544b93f35c3c088eae29e0d8":{"kind":"REWORK","_number":2,"created":"2019-08-20 18:37:39.000000000","uploader":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"ref":"refs/changes/37/677537/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/37/677537/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/37/677537/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/37/677537/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/37/677537/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/37/677537/2"}}},"commit":{"parents":[{"commit":"291489176cecf7e087a3964680db21b037cfc672","subject":"Fix publish path for api-jobs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/291489176cecf7e087a3964680db21b037cfc672"}]}],"author":{"name":"Andreas Jaeger","email":"aj@suse.com","date":"2019-08-20 18:30:32.000000000","tz":120},"committer":{"name":"Andreas Jaeger","email":"aj@suse.com","date":"2019-08-20 18:37:30.000000000","tz":120},"subject":"Add more promote docs jobs","message":"Add more promote docs jobs\n\nAdd promote-openstack-tox-docs-direct job for publishing jobs without\n/latest.\n\nAdd promote-openstack-tox-docs-infra job for publishing infra repos.\nConvert osf/openstackid to use it and remove release job, the infra job\ndoes not handle releases extra.\n\nChange-Id: I0da6c940b59a66feaf87e353ba53d473a35122f4\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/da91ffe986ed89b9544b93f35c3c088eae29e0d8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/da91ffe986ed89b9544b93f35c3c088eae29e0d8"}]},"branch":"refs/heads/master"},"6b89ce404df7d4a484d5c2972105ac1bc025be4e":{"kind":"REWORK","_number":3,"created":"2019-08-20 18:40:53.000000000","uploader":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"ref":"refs/changes/37/677537/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/37/677537/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/37/677537/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/37/677537/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/37/677537/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/37/677537/3"}}},"commit":{"parents":[{"commit":"291489176cecf7e087a3964680db21b037cfc672","subject":"Fix publish path for api-jobs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/291489176cecf7e087a3964680db21b037cfc672"}]}],"author":{"name":"Andreas Jaeger","email":"aj@suse.com","date":"2019-08-20 18:30:32.000000000","tz":120},"committer":{"name":"Andreas Jaeger","email":"aj@suse.com","date":"2019-08-20 18:40:34.000000000","tz":120},"subject":"Add more promote docs jobs","message":"Add more promote docs jobs\n\nAdd promote-openstack-tox-docs-direct job for publishing jobs without\n/latest.\n\nAdd promote-tox-docs-infra job for publishing infra repos.\nConvert osf/openstackid to use it and remove release job, the infra job\ndoes not handle releases extra.\n\nChange-Id: I0da6c940b59a66feaf87e353ba53d473a35122f4\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/6b89ce404df7d4a484d5c2972105ac1bc025be4e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/6b89ce404df7d4a484d5c2972105ac1bc025be4e"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
